)]}'
{"id":"openstack%2Fpuppet-heat~836233","triplet_id":"openstack%2Fpuppet-heat~master~Icdf4f0201dd1e5f93a450473709851977ec20034","project":"openstack/puppet-heat","branch":"master","topic":"clients-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-05 15:39:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icdf4f0201dd1e5f93a450473709851977ec20034","subject":"Support more clients configurations","status":"MERGED","created":"2022-04-02 17:18:32.000000000","updated":"2022-04-05 15:41:04.000000000","submitted":"2022-04-05 15:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836233-clients-opts","meta_rev_id":"dc06e39d16e8c37e21400f0d485a3c29f901363c","_number":836233,"virtual_id_number":836233,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 15:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-04-04 19:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-05 13:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-04-05 13:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-02 18:31:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 19:58:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-04-05 13:58:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f5f2064454d4fa811d6c8f0cf20478d0f4daf48e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-02 17:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e721ab7baaf71ec4c3c862ba0291c8e9082b4c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 18:31:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7032289b495423f8f6c0630792e881d : FAILURE in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/917c917b826b4bf1bc4a3404095f16e6 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5d0a3b8a6b64637a3d66d81d37a77ae : SUCCESS in 2m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad31c55030dc416ca93a2de812a306df : FAILURE in 10m 53s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a40d893c009c40898985edb3ad46614f : FAILURE in 10m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96bbb489e194f809b23d935e3e10344 : FAILURE in 8m 32s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bab39c1947bd468bb2d3e65bc791a97a : FAILURE in 9m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dbbcbe256a64ec995a825f102d64609 : FAILURE in 8m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45763970cfb94f24ac07ba7589f1ce23 : FAILURE in 26m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1df7cf49bab4d63add2941bd34d648e : SUCCESS in 49m 43s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46c9ae7a84574340b0d28d6121ee3ba0 : SUCCESS in 48m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921e0d3a289548e0bf66164562b22317 : SUCCESS in 13m 20s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3903c1c1cd884df3a5d7e9349c09fb90 : SUCCESS in 18m 42s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/59bd177622ef4c69a5e4318d188f2db5 : FAILURE in 6m 33s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3777944186724b129a542f2990ddf7b1 : FAILURE in 11m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5ab6fa2d76c4f55a756fe1fb8ceed3d : FAILURE in 12m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cb2bdb7d8604f9bb2a9345b29ed339d : SUCCESS in 4m 29s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a090b216faf342dc9b7d76b37d17c088 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a2b832428de47034e5d36799bfa1745581448af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 05:14:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16bbbc686520751fe97772bc1aae635cc2129859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 06:33:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb793c189b4347e9adaf4e98f3095ef5 : FAILURE in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d0e666423c743b199fbb12987f25638 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/772b2bfe6f2e4cd8a63c2581a62af88b : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7b3983de1ba4b4f8f1aba299f1d0a01 : FAILURE in 13m 03s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/29be5e0f45a64d299a8bd51b7cbadec3 : FAILURE in 13m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42b8116240314d1392b206a272626587 : FAILURE in 10m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53666e51c7d4b0aaa2cf04a59b60ad4 : FAILURE in 10m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1181ffe9357446df9b99fa848879c3b7 : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2f74e07afe34297b4efbb4a0def57c7 : FAILURE in 42m 04s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca82a3cc641c4c3e9fe0ac225b20a836 : SUCCESS in 1h 06m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f607bc61894a4a8ea9e0f17c99debc3e : SUCCESS in 46m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0680a6e494d0459e8fafc65b55e26cc6 : SUCCESS in 15m 31s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0be9484028c34bdc820aac6f3f647771 : SUCCESS in 21m 18s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b37b1a4fd4f24ebd8001ed49ea60271a : FAILURE in 6m 57s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51e4d56785fa4641800f4631cca14a66 : SUCCESS in 18m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df3bfc608b24461bc08d09b8943e8e4 : SUCCESS in 16m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4f6b415653f422aacf2756b606b0fe8 : SUCCESS in 3m 34s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8619dde82a5f48cf843e19f66e5579c3 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":2},{"id":"198d541d953cded0ee978eda6403f5c7359cdcf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 11:17:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae57af35b179255be1795bfdda5ede766c31bb28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 12:40:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42f5d6d743184c32b250ddbb28d6c253 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa0833688d98479686ae473cb8755fc0 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9988deb032634c5da1e81eb806782d8c : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b500b40d88a54ccdab856c7b7468e2b4 : FAILURE in 10m 42s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2dc49dcad1794f40bd7192dea64cf387 : FAILURE in 10m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a05820477a1488e9c83661fd9d761e5 : FAILURE in 9m 52s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f12b52e5ae204886955334edd51ff749 : FAILURE in 8m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55648c97db044b238df5f63dfefab8de : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae033ebf7bf54b028675a09bca9cc0d7 : FAILURE in 26m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a76555ccfeb949adbd5fbe3b95f5069a : SUCCESS in 51m 40s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/829f4d9b9ccb4acf9b63c081f522414e : SUCCESS in 45m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81625623d4084e2d9c3872f4250265dd : SUCCESS in 13m 56s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6c9a46b450f49688c5a33b0caff68b6 : SUCCESS in 22m 46s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce1e7516aab0459eaa289a2fe8e577df : FAILURE in 9m 22s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93480ab3904a4f059c49410011a3d8bb : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a41359c744164ef3966eaf7a0051129e : FAILURE in 12m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f301bce11f8040208bf36d62964976b2 : SUCCESS in 3m 15s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c79d008dcd1949dcbd4bdecbeb34ec7a : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":3},{"id":"26f30c117f4852b4d3df52d7fc67e70dfe739dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 12:57:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6b56b8b53be643b10ceb3d660d4cd8ead503f1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 14:19:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/443aba83ce8b4a0da8e456df6508c93d : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7833bfa973f4ee9ad5e15e61812605a : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e93fc938f6a7494a9af70ee6929cd666 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2e44dde3377442fa0d599f8dcfaf089 : FAILURE in 12m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f43a23ea61f34d0f82ffa1b5773d3e03 : FAILURE in 11m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a37071be5b204ec49f54dd9de64b76ec : FAILURE in 10m 52s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e44e7d2fdbb54f6ca12dbaf961cd7524 : FAILURE in 9m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13db0e4ea3f64eacb8856bf3229f844a : FAILURE in 10m 37s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9b820c71e9742e596a23f81f4da23d0 : FAILURE in 39m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/70dea9f290e84fe2b19583f88e767f2a : SUCCESS in 1h 10m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15f3b477e58044d1ae73c3371e83c390 : SUCCESS in 56m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45122e1788624d248858d7bd4711b048 : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4fcac3e76064ef79d1040c8b4ff26e1 : SUCCESS in 18m 08s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/86870d0bde2a4009a6e96cffff4a125f : FAILURE in 7m 54s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6f11160d0a4c19b94404fa49054b21 : FAILURE in 13m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1847d1e057a24dc2a62a609f5b08a988 : SUCCESS in 16m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/808a92d57c5845d79d0fd84df7c58e79 : SUCCESS in 3m 30s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/edf793ea06b04a43ab59afb33ac24cfb : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ae718e131f83b90d5062696ff7d5ee5be965bd88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-03 14:22:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6ac06e539dd3d32db083d143f2418289d5c25b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 15:47:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fc2274d4d0d43db8d91ec3e42a9975c : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df6965d67a4248eb99d65bcf0c213096 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1014d2f071164355bd2296f208970071 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a1c914bfbd34f2daaae4d655aeedc00 : SUCCESS in 12m 07s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea1692e47f384dbd8c745250c058c83d : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57cbd5bc9a5f4a7b9eff7b99d33c058f : SUCCESS in 10m 57s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f67843d273db4d7db1a913beec29427a : SUCCESS in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bdf816ce56a4676b596a109cce513e2 : SUCCESS in 9m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dda67c4c9dc74c7b89f311ceaf408190 : FAILURE in 39m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/28943014f35e4cd399defbc22228937a : SUCCESS in 1h 05m 09s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51534ff9d8424c3bb963a68c0c57de0a : SUCCESS in 56m 48s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c839d4e39af4407c8e75bd79d2d54db6 : SUCCESS in 16m 09s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef1e1a730f374d179d19bc64d62e6027 : SUCCESS in 20m 16s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5484820141443e09d084e9d15c5cdd4 : FAILURE in 8m 43s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c96faa4a8a348e9a7c7a3d638f59f1c : FAILURE in 14m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f4eb05ae01d40cc8b033a1866780654 : FAILURE in 12m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb6bc8e6b7a64e61ba898eebc06fc492 : SUCCESS in 3m 40s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d5d42188b41b4320b6e657c0c04581a9 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":5},{"id":"adad8eec280907fcc3c26ba8047505e2b45abf1b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-04 19:58:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"aa5289000ffbe98f7ef767675929beaa86d9b9ec","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-04 19:58:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4b684854bd194e6923a5e36ff1675e1bfe8c9114","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-04-05 13:58:25.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ccae01ace4947de2223bad725f95f292d7e1de82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:00:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7742d17e1f4c55beff1a1c0a14c06151426de0c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:39:00.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f8f82a7551b4d699763ca0cc1d620c1 : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d984273f0b77401b9713b0152f56630a : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea9e13e35bd2490cbb91b2f0aae5afbe : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ebb9a18521b44d8966ff9831366de9d : SUCCESS in 15m 58s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7143cd697f9248fab5ff989bfad8d26f : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd165bd60dc647e6a75f77a78aea71a4 : SUCCESS in 11m 35s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/833b67f850d148328e24ebee6ed456c7 : SUCCESS in 12m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd89f95b3a7a42748580fc0ac24fec08 : SUCCESS in 1h 11m 51s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0ec2492888f4a638d96e9051c397057 : SUCCESS in 25m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ef99b0b3dc4802baaaf206aea73bfd : SUCCESS in 4m 12s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/966e4f382c4d4577aeb6e1fa5b4564b0 : SUCCESS in 1h 37m 06s","accounts_in_message":[],"_revision_number":5},{"id":"79f9db00bcf321d8bcaf52d6201eba30113def4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:39:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"dc06e39d16e8c37e21400f0d485a3c29f901363c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:41:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81dc5de2db834febbd2f7354af15edc6 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0d043d793fa9eb83e87cb008b073329bb063bfb1","revisions":{"b29551364f619d17b210177e23d54bdc962968ec":{"kind":"REWORK","_number":1,"created":"2022-04-02 17:18:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/836233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/836233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/1"}}},"commit":{"parents":[{"commit":"e7cefc1ae06ba8db5c04ad84df403edd1435683e","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7cefc1ae06ba8db5c04ad84df403edd1435683e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:18:22.000000000","tz":540},"subject":"Support more clients configurations","message":"Support more clients configurations\n\nThis change introduces support for [clients] parameters and\n[clients_\u003cservice\u003e] parameters, which determine behavior of service\nclients used by heat for communication with the OpenStack services.\n\n- Implementation of heat::clients::\u003cservice\u003e are generally same.\n  Only cinder, heat, keystone and nova provides additional parameters.\n\n- The existing parameters in the base heat class are deprecated in\n  favor of the new classes.\n\nChange-Id: Icdf4f0201dd1e5f93a450473709851977ec20034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b29551364f619d17b210177e23d54bdc962968ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b29551364f619d17b210177e23d54bdc962968ec"}]},"branch":"refs/heads/master"},"4fa6d703caf569d8978b5f1da131c74c8389bb05":{"kind":"REWORK","_number":2,"created":"2022-04-03 05:14:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/836233/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/836233/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/2"}}},"commit":{"parents":[{"commit":"e7cefc1ae06ba8db5c04ad84df403edd1435683e","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7cefc1ae06ba8db5c04ad84df403edd1435683e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 05:14:11.000000000","tz":540},"subject":"Support more clients configurations","message":"Support more clients configurations\n\nThis change introduces support for [clients] parameters and\n[clients_\u003cservice\u003e] parameters, which determine behavior of service\nclients used by heat for communication with the OpenStack services.\n\n- Implementation of heat::clients::\u003cservice\u003e are generally same.\n  Only cinder, heat, keystone and nova provides additional parameters.\n\n- The existing parameters in the base heat class are deprecated in\n  favor of the new classes.\n\nChange-Id: Icdf4f0201dd1e5f93a450473709851977ec20034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4fa6d703caf569d8978b5f1da131c74c8389bb05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4fa6d703caf569d8978b5f1da131c74c8389bb05"}]},"branch":"refs/heads/master"},"2dd6e7626d795983ec3ac7043cfed69d0ab1cb81":{"kind":"REWORK","_number":3,"created":"2022-04-03 11:17:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/836233/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/836233/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/3"}}},"commit":{"parents":[{"commit":"e7cefc1ae06ba8db5c04ad84df403edd1435683e","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7cefc1ae06ba8db5c04ad84df403edd1435683e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 11:17:31.000000000","tz":540},"subject":"Support more clients configurations","message":"Support more clients configurations\n\nThis change introduces support for [clients] parameters and\n[clients_\u003cservice\u003e] parameters, which determine behavior of service\nclients used by heat for communication with the OpenStack services.\n\n- Implementation of heat::clients::\u003cservice\u003e are generally same.\n  Only cinder, heat, keystone and nova provides additional parameters.\n\n- The existing parameters in the base heat class are deprecated in\n  favor of the new classes.\n\nChange-Id: Icdf4f0201dd1e5f93a450473709851977ec20034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/2dd6e7626d795983ec3ac7043cfed69d0ab1cb81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/2dd6e7626d795983ec3ac7043cfed69d0ab1cb81"}]},"branch":"refs/heads/master"},"aaa91dddd2f461801fcdcc3965d066c569962721":{"kind":"REWORK","_number":4,"created":"2022-04-03 12:57:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/836233/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/836233/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/4"}}},"commit":{"parents":[{"commit":"e7cefc1ae06ba8db5c04ad84df403edd1435683e","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7cefc1ae06ba8db5c04ad84df403edd1435683e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 12:57:47.000000000","tz":540},"subject":"Support more clients configurations","message":"Support more clients configurations\n\nThis change introduces support for [clients] parameters and\n[clients_\u003cservice\u003e] parameters, which determine behavior of service\nclients used by heat for communication with the OpenStack services.\n\n- Implementation of heat::clients::\u003cservice\u003e are generally same.\n  Only cinder, heat, keystone and nova provides additional parameters.\n\n- The existing parameters in the base heat class are deprecated in\n  favor of the new classes.\n\nChange-Id: Icdf4f0201dd1e5f93a450473709851977ec20034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/aaa91dddd2f461801fcdcc3965d066c569962721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/aaa91dddd2f461801fcdcc3965d066c569962721"}]},"branch":"refs/heads/master"},"0d043d793fa9eb83e87cb008b073329bb063bfb1":{"kind":"REWORK","_number":5,"created":"2022-04-03 14:22:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/836233/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/33/836233/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/33/836233/5"}}},"commit":{"parents":[{"commit":"e7cefc1ae06ba8db5c04ad84df403edd1435683e","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e7cefc1ae06ba8db5c04ad84df403edd1435683e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-02 17:06:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-03 14:22:26.000000000","tz":540},"subject":"Support more clients configurations","message":"Support more clients configurations\n\nThis change introduces support for [clients] parameters and\n[clients_\u003cservice\u003e] parameters, which determine behavior of service\nclients used by heat for communication with the OpenStack services.\n\n- Implementation of heat::clients::\u003cservice\u003e are generally same.\n  Only cinder, heat, keystone and nova provides additional parameters.\n\n- The existing parameters in the base heat class are deprecated in\n  favor of the new classes.\n\nChange-Id: Icdf4f0201dd1e5f93a450473709851977ec20034\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0d043d793fa9eb83e87cb008b073329bb063bfb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0d043d793fa9eb83e87cb008b073329bb063bfb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
