)]}'
{"id":"openstack%2Fpuppet-heat~841652","triplet_id":"openstack%2Fpuppet-heat~master~I3806ab51e6593a9e00172af6b1eb02d9812fcf72","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-24 10:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3806ab51e6593a9e00172af6b1eb02d9812fcf72","subject":"Remove deprecated heat::max_stack_per_tenant","status":"MERGED","created":"2022-05-12 23:51:42.000000000","updated":"2022-05-24 10:38:38.000000000","submitted":"2022-05-24 10:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841652","meta_rev_id":"9018d8488b546faf512ef0a4dbe18567fbd87d50","_number":841652,"virtual_id_number":841652,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-24 10:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-17 16:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-24 08:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-24 08:59:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 01:20:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 16:46:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-24 08:59:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"b1f25f117af80fc063e9833865c3650e30e4c61d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-12 23:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fee3d752804db8b7b758f953fe8b1136dc264ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 01:20:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85437e3bcc7641ef81bddc7ab93c17ed : FAILURE in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/648ed677f9a54d71b5b844ae657c2203 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8486d073a74498d8347e74932d5572d : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/31ab54f7827d40dc9bedf5b36e0dad35 : FAILURE in 13m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5315d742cec54ac1a99d85a3fe34ff88 : FAILURE in 9m 34s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3cc92a21ee444e2882bd04d0aedf652 : FAILURE in 9m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe18f6164c4d46a19b3c6b6ea792d605 : FAILURE in 8m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bba7135fc56c4bfeb513f49983712fa1 : FAILURE in 34m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/092a2d4506714712bccb3d999e1c35f8 : SUCCESS in 45m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceb5637989f14bf3a7a5ffb351d1c57a : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ed8ab5a8f444b33b5775b8d1ddd3879 : SUCCESS in 19m 27s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5779a98c399455db344399850f0c469 : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcbce440f2ba4ec9a4c0fac9f034263d : SUCCESS in 14m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/197c5c2dceb8465f9a0b187e71f0550d : SUCCESS in 14m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794a096b4ab54bcda22eef9da7efe446 : SUCCESS in 3m 08s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9c2c555721c46b181e7e0234c245d86 : SUCCESS in 1h 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9c18288802a711a863731fcba87cde4cf137c48a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-13 01:40:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10de0e660421ccd366dcafc8d4b64938519ee2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 03:24:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8115067a6334e58adb208f4a29be30a : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dcb01d4253347368339bfb60cd0a718 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acbe8bd5fda4068bb624de378fa12d4 : SUCCESS in 2m 54s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9135fe23ab184daab1866b582d1ddb10 : SUCCESS in 11m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/889630fb8bf5498eac25d05866f2c98d : SUCCESS in 9m 23s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5caa8a51492940d0a49b602c269109b6 : SUCCESS in 10m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cc86a174fdc41ecb7897325a547c2b1 : SUCCESS in 9m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de8f41ad34d84f1780732b3591fb479f : FAILURE in 30m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a5d56bf63dc4a878e43f1a435c31f25 : SUCCESS in 1h 04m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64b7b872811c4e3f9cd4781db26d9547 : SUCCESS in 16m 06s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c3960ce4b9044cc86bddaaf09474a44 : SUCCESS in 20m 47s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/be6454b2bfcf48358441d2320903e34f : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d2350eb0bba4b909971c8131047eda3 : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e89bf21a036541e09e668d71d0131932 : SUCCESS in 15m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/213f8a8115ba47738b32b0f3d2a152ea : SUCCESS in 3m 21s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a37ac282a0184855a222ae89b2660cbf : SUCCESS in 1h 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"923bbf7c0e0810e72281695b069f0be10f17147c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-17 16:46:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49fcec340f28682fbd0671ca6a02c450e8fd23ea","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-24 08:59:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f4c740336fb772df0b145d16512a68144f099cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 09:00:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ea34caac3d480d19f634b90bc605ffd93aeffc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:37:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/195a5232d96d4c3d901f6636dadaf2de : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9187da4cbe154d359aaca8b664895a4c : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abb023e0922d43bc9b13989f2efd68f5 : SUCCESS in 3m 34s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/09e29cb128ca4d6ba4a28653505c00ef : SUCCESS in 15m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf26f8338f12450e832815bef5259e04 : SUCCESS in 12m 07s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d9d98ff0e954351a8ea6161c11716e4 : SUCCESS in 11m 19s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b54c4b818b341a58774fcce91ca896f : SUCCESS in 1h 06m 28s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a3c1068737a4b7197930ba1fdf06e78 : SUCCESS in 24m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d65b383a4132440184faae9c633b4589 : SUCCESS in 4m 08s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e04187a1a9d04be28491a952a109e8fb : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c9c444cff19f3d30773647ddcb6f36766baf151f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9018d8488b546faf512ef0a4dbe18567fbd87d50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:38:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fdc2fe8eaba450d8086e81d6f76c69d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68ec6e99d3d515ba203c8c1d1784daeea4825833","revisions":{"1b1c025625222cf0540ccae582bf05c2941fd7a1":{"kind":"REWORK","_number":1,"created":"2022-05-12 23:51:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/841652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/841652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/1"}}},"commit":{"parents":[{"commit":"058565693733bb058af2e627b8999cc7acefcc75","subject":"apache+mod_wsgi: Disable SSL by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/058565693733bb058af2e627b8999cc7acefcc75"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-12 23:51:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-12 23:51:19.000000000","tz":540},"subject":"Remove deprecated heat::max_stack_per_tenant","message":"Remove deprecated heat::max_stack_per_tenant\n\n... because it was deprecated during Xena cycle by [1].\n\n[1] e79e74cd089622bc33ebcc0ceb2bb17c398e216f\n\nChange-Id: I3806ab51e6593a9e00172af6b1eb02d9812fcf72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1b1c025625222cf0540ccae582bf05c2941fd7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1b1c025625222cf0540ccae582bf05c2941fd7a1"}]},"branch":"refs/heads/master"},"68ec6e99d3d515ba203c8c1d1784daeea4825833":{"kind":"REWORK","_number":2,"created":"2022-05-13 01:40:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/841652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/52/841652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/52/841652/2"}}},"commit":{"parents":[{"commit":"058565693733bb058af2e627b8999cc7acefcc75","subject":"apache+mod_wsgi: Disable SSL by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/058565693733bb058af2e627b8999cc7acefcc75"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-12 23:51:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-13 01:40:21.000000000","tz":540},"subject":"Remove deprecated heat::max_stack_per_tenant","message":"Remove deprecated heat::max_stack_per_tenant\n\n... because it was deprecated during Xena cycle by [1].\n\n[1] e79e74cd089622bc33ebcc0ceb2bb17c398e216f\n\nChange-Id: I3806ab51e6593a9e00172af6b1eb02d9812fcf72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/68ec6e99d3d515ba203c8c1d1784daeea4825833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/68ec6e99d3d515ba203c8c1d1784daeea4825833"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
