)]}'
{"id":"openstack%2Fpuppet-heat~849172","triplet_id":"openstack%2Fpuppet-heat~master~I5af2205b0e067f1a4fe44b4d4fd92bf23fd24de4","project":"openstack/puppet-heat","branch":"master","topic":"bug/1981118","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-15 20:50:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5af2205b0e067f1a4fe44b4d4fd92bf23fd24de4","subject":"Switch provider to manage paste.ini file","status":"MERGED","created":"2022-07-09 04:09:03.000000000","updated":"2022-07-15 20:50:33.000000000","submitted":"2022-07-15 20:50:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849172-bug/1981118","meta_rev_id":"de2031d2bba47a5320f4355562105a96c3ff962b","_number":849172,"virtual_id_number":849172,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 21:53:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-07-15 19:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-07-15 19:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 05:58:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 21:53:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-15 19:15:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4352e3025d2d499a0d281b43b8822ad6d780398a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:09:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a4b41f0225e37fc71ff9846c19fe9f41cc02058","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:09:28.000000000","message":"Topic config-acceptance removed","accounts_in_message":[],"_revision_number":1},{"id":"731ac97ad5cb4beea901006dd02e9639230169ec","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 04:09:31.000000000","message":"Topic set to paste-ini-provider","accounts_in_message":[],"_revision_number":1},{"id":"b63ffbcd559818f51d3357a397602d76f7ea6070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 05:58:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd206fca9e2f4f688ced7b6b92102f51\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a33934df13f74162ac4a557bb75b7cbf : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c800be06d57d423b88ab80f2324b1aa3 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dba2d1c789d4422a3a17a1c9bda22cd : SUCCESS in 2m 52s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dea804f52194350b69898b8e754f94e : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1503ebba722f43aeaff2d305963506d9 : SUCCESS in 9m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/029445dd9bb94582b0de10c9428c43ce : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/790ca8f8666d4cada7c1c07794b2d650 : SUCCESS in 8m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4941f00e87c245ae9f89866f7b7c216b : FAILURE in 37m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5233838c7cdb4def8fe8a1dbb6a8e65d : SUCCESS in 1h 01m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14c56e0af5f4a8fba3fd4acb0d54f0c : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b8c035df40a4c919abd1dd89624ef63 : SUCCESS in 19m 44s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b82d03532eb4158b6200861b4ea7d91 : SUCCESS in 14m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1d86c80387c4d3b94bfb38f5fd02083 : SUCCESS in 15m 10s (non-voting)\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c1b7138f66ef43bdb04625cf11c441e8 : SUCCESS in 1h 44m 57s","accounts_in_message":[],"_revision_number":1},{"id":"abc6c82553906ed25ece44b2b6f487c2ee932a0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:04:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bf419d1e52c174dcfcf5191ece864741636630b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:04:27.000000000","message":"Topic paste-ini-provider removed","accounts_in_message":[],"_revision_number":2},{"id":"76fe901ad3d037373d3a50555c6472e0347b739f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 10:04:30.000000000","message":"Topic set to bug/1981118","accounts_in_message":[],"_revision_number":2},{"id":"01dbd612f891505bd62c04b0bcd8a25fe817af7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 11:23:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432db47a4bc84456b568cbf535d702f4\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7185035246734dc9a2bd8f65d9f74cbf : SUCCESS in 2m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e659815febb474f86c510cfdef79bbc : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b70e04db3b944cbba0fb545eae4e5950 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c067f35cab9a4cb5ba13bb8248a47e1d : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14325fbae1e146e8896cdd6e6e874eb1 : SUCCESS in 8m 05s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2864652257714fa084e527c82838b670 : SUCCESS in 10m 35s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f543c0433884d66874e017d4b1a1c26 : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35f0ccff97b8494aa0ffe8d5802ffe55 : FAILURE in 32m 09s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa345658c01e476997314cc6c233afdd : SUCCESS in 51m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf4f4c60692945f2beb8f2aef674c192 : SUCCESS in 16m 37s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/704b04e6cae04d8fa29a080921124ff7 : SUCCESS in 17m 01s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a00b754c4534432fbddceeb7b4730912 : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c02228d7a3294a5789fc12081e1ea4a5 : SUCCESS in 15m 35s (non-voting)\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c66f51813cf74861848e3f8d64062b56 : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":2},{"id":"58b3d279b25b7e0f02a2570117332869ee869cb9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 21:53:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27a27dc921692c808f6b7dc6d58046bf9435ee56","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:15:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e940a79f4878b078d1bf8903bc1f1159a3b2b687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:16:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cf093bb3fccea304653fd36622b0e5a0a0bedcf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:50:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2eb9d43e504e98a97cafaa0293e024\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77f14b8f0b53438588a3ed5366856879 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59d942cd21734efabe52906205e89966 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6904910ac26f49c3adce4e76343dfc34 : SUCCESS in 5m 18s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4564340a42254d69b246618f88457431 : SUCCESS in 24m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e640117dd6e14b52bccc185619f01a9b : SUCCESS in 19m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf5bc414929d4091818f14ff389a316e : SUCCESS in 19m 29s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b04a181cd6b84b1b879976cf332e5ac8 : SUCCESS in 1h 05m 40s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/87f65142a28d4c5d93eb8e6d0ffa3934 : SUCCESS in 31m 37s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/809aa8c60f1647c6bfa76ac10959e4bc : SUCCESS in 1h 27m 55s","accounts_in_message":[],"_revision_number":2},{"id":"de2031d2bba47a5320f4355562105a96c3ff962b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:50:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0daf2873414c8b26aad2c3c8c550963d60ec9ea0","revisions":{"60221a49189f2caebdf2e37901b8b29d988d6a82":{"kind":"REWORK","_number":1,"created":"2022-07-09 04:09:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/849172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/72/849172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/1"}}},"commit":{"parents":[{"commit":"98c92fe15ce0e0557e85fff20ecf809f7607ef54","subject":"Ensure [DEFAULT] deferred_auth_method is purged","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/98c92fe15ce0e0557e85fff20ecf809f7607ef54"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:08:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:08:55.000000000","tz":540},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nChange-Id: I5af2205b0e067f1a4fe44b4d4fd92bf23fd24de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/60221a49189f2caebdf2e37901b8b29d988d6a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/60221a49189f2caebdf2e37901b8b29d988d6a82"}]},"branch":"refs/heads/master"},"0daf2873414c8b26aad2c3c8c550963d60ec9ea0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-09 10:04:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/849172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/72/849172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/72/849172/2"}}},"commit":{"parents":[{"commit":"98c92fe15ce0e0557e85fff20ecf809f7607ef54","subject":"Ensure [DEFAULT] deferred_auth_method is purged","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/98c92fe15ce0e0557e85fff20ecf809f7607ef54"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 04:08:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 10:04:14.000000000","tz":0},"subject":"Switch provider to manage paste.ini file","message":"Switch provider to manage paste.ini file\n\n... so that we users can use \u0027\u003cSERVICE DEFAULT\u003e\u0027 similarly to\nthe resource types to manage the \u003cservice\u003e.conf files.\n\nCloses-Bug: #1981118\nChange-Id: I5af2205b0e067f1a4fe44b4d4fd92bf23fd24de4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0daf2873414c8b26aad2c3c8c550963d60ec9ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0daf2873414c8b26aad2c3c8c550963d60ec9ea0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
