)]}'
{"id":"openstack%2Fpuppet-heat~876489","triplet_id":"openstack%2Fpuppet-heat~master~If4ea3913c5ec9c3719d57b8b682a87816cbaabf7","project":"openstack/puppet-heat","branch":"master","topic":"executor_thread_pool_size","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-06 17:14:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4ea3913c5ec9c3719d57b8b682a87816cbaabf7","subject":"Expose executor_thread_pool_size","status":"MERGED","created":"2023-03-06 04:41:22.000000000","updated":"2023-03-06 17:15:25.000000000","submitted":"2023-03-06 17:14:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876489-executor_thread_pool_size","meta_rev_id":"e0be4fbcc0073c3bd0eb2390ac2105500a14efa6","_number":876489,"virtual_id_number":876489,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-06 17:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-06 07:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-06 15:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-06 15:21:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 06:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 07:14:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-06 15:21:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3234dfabf654ee8afa219c52d0b705302ec8257c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-06 04:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76400551d9af1369d523f5e7b0a8e3776a30403e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-06 04:47:38.000000000","message":"Topic set to executor_thread_pool_size","accounts_in_message":[],"_revision_number":1},{"id":"454e6be778f0367c2329532e318b884bbdd1cc2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-06 04:48:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77d0895c579855d0a79467021134e233d14a94a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 06:27:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b0abbdf31e542f7b8e483d407b2fc16\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab9027dfea824030b67f034665a594e4 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01cda73d0f724234a5189daa9ed34cbb : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c6a3f8bffc6418cb8bf48bf3d20b493 : SUCCESS in 12m 41s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6ab1d126f474748aa773b20d57a0dbe : SUCCESS in 9m 18s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5588ceef47349629cba4e4cc5797878 : SUCCESS in 9m 27s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/850d8f9bfedc4b848a7473a39d16082e : FAILURE in 27m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abb6547b22724ea4a00bf9cb4c25925e : SUCCESS in 56m 06s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/68c44a31ca06433a94409dc6f8811e3d : SUCCESS in 21m 49s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b79f96885e9454fa21396909c8974f2 : SUCCESS in 15m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/edc5af90e6b1453eb6269e7b4b1fa912 : SUCCESS in 16m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b54189eb26134936949f5c92122e637f : SUCCESS in 3m 39s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aaf4994a6d2d4f869d34d68a99a9f0cc : SUCCESS in 1h 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7dca0271a614cd79c73ba4cf0186d0d66d71d5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-06 07:14:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"34bed1c2a57d3c65d82ab4baf8a2ce4bdc09ac2b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-06 15:21:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e012e655e0224c155d11d92c3524bfeb6389bdde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:22:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9c7c25972ed66ffe98a034a174752ecd85dc78f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 17:14:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc37035e32a4eeb93d89c6086c0258a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f4042b6032f4a369b57751c161bd8b5 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af8314bd2ec24d549bbf921f613c5b84 : SUCCESS in 2m 20s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8768b5f389024ac494049e7929bea65b : SUCCESS in 10m 43s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e1fe39c3a84e7d86a22415d267bb0c : SUCCESS in 7m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/73a91a131b0b47419feef5224f05cf72 : SUCCESS in 1h 03m 07s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1c6a01a4fab4f9bb5b51b495ee442de : SUCCESS in 19m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99719658239b43bdb9655bd8acd1cee6 : SUCCESS in 2m 42s\n- puppet-heat-tripleo-standalone https://zuul.opendev.org/t/openstack/build/172673ff42384fa88e6d09e63b94849e : SUCCESS in 1h 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e56e10bcf16308ad525350f6e85efa0b38ba3cf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 17:14:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e0be4fbcc0073c3bd0eb2390ac2105500a14efa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 17:15:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb3b402d75a476687901695ce40b6a4\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75508a4131154ebca34ed6ac72c6c223 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee09225ba9ea1d68f79dbadd850a878c671c685b","revisions":{"887aa2ef4e548a6b585488e2add3a35741232685":{"kind":"REWORK","_number":1,"created":"2023-03-06 04:41:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/876489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/89/876489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/1"}}},"commit":{"parents":[{"commit":"da45ce8a21de067b8abd598f7ab41de9911185ae","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/da45ce8a21de067b8abd598f7ab41de9911185ae"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 04:40:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 04:40:28.000000000","tz":540},"subject":"Expose executor_thread_pool_size","message":"Expose executor_thread_pool_size\n\nThis option has been supported by puppet-oslo but has not been\nconfigurable.\n\nChange-Id: If4ea3913c5ec9c3719d57b8b682a87816cbaabf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/887aa2ef4e548a6b585488e2add3a35741232685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/887aa2ef4e548a6b585488e2add3a35741232685"}]},"branch":"refs/heads/master"},"ee09225ba9ea1d68f79dbadd850a878c671c685b":{"kind":"REWORK","_number":2,"created":"2023-03-06 04:48:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/876489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/89/876489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/89/876489/2"}}},"commit":{"parents":[{"commit":"da45ce8a21de067b8abd598f7ab41de9911185ae","subject":"Replace legacy facts and use fact hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/da45ce8a21de067b8abd598f7ab41de9911185ae"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 04:40:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 04:48:27.000000000","tz":540},"subject":"Expose executor_thread_pool_size","message":"Expose executor_thread_pool_size\n\nThis option has been supported by puppet-oslo but has not been\nconfigurable.\n\nChange-Id: If4ea3913c5ec9c3719d57b8b682a87816cbaabf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ee09225ba9ea1d68f79dbadd850a878c671c685b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ee09225ba9ea1d68f79dbadd850a878c671c685b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
