)]}'
{"id":"openstack%2Fpuppet-heat~879122","triplet_id":"openstack%2Fpuppet-heat~stable%2Ftrain~If4ea3913c5ec9c3719d57b8b682a87816cbaabf7","project":"openstack/puppet-heat","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-31 17:04:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4ea3913c5ec9c3719d57b8b682a87816cbaabf7","subject":"Expose executor_thread_pool_size","status":"MERGED","created":"2023-03-31 04:26:18.000000000","updated":"2023-03-31 17:05:30.000000000","submitted":"2023-03-31 17:04:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879122","meta_rev_id":"5a8661ab5b5f815c3072083019606b7adec4edac","_number":879122,"virtual_id_number":879122,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-31 17:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-03-31 09:04:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-03-31 14:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-03-31 14:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 08:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-31 09:04:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-03-31 14:22:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"7376e88bf990eda097c5a98738c8af6120528043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-31 04:26:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3fd85b8457c02caa8db80ffd6225608c496ce3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 08:07:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33bd43a860104e9785d73be56143c5e0\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b71de0f91704624aad36a7530365631 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bf91f4bd69e4ab0bc393b050c7c6f71 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b122399a51644699cd7abb0afbb26e0 : SUCCESS in 3m 49s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d09b206461be4501b2644b4ef0ee04a7 : SUCCESS in 20m 53s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/96339a4bdd32448bad33c776b4735960 : SUCCESS in 11m 50s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6230e93f840499caba3f85451e331bb : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db6179b0781541c18aad35487b73de47 : SUCCESS in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/605bdb75001b47d0acceb5c9b91904e8 : SUCCESS in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/980d5fe9693a4def8f7265e7395406b1 : SUCCESS in 38m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d167bc47bb8c4eb1811cf9f24ec7a783 : SUCCESS in 1h 12m 19s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4455609093614cd89c3b5c544720eff2 : SUCCESS in 56m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/44ab0456fddd4e9da9c1c82b10c95ff6 : SUCCESS in 18m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87897e989bd440578276ecbdddbca925 : FAILURE in 6m 14s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/410ef51676c441ee97c4933076b7c13e : SUCCESS in 3h 33m 54s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e64ba9d95744702920ea824e1c4d6a0 : SUCCESS in 1h 55m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ced09cfcaccc41c5ac56a367a1a7efb1 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"dfecbd59db9d85b293f822993999370eb39fea3b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-31 09:04:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce38ed2d118d2b177b7339b6e8620740f86f09cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-03-31 14:22:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e46ac853c946cf8e61b74c0913c4542800872ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 14:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bd166bbfccd61b491e6c7a879c544693efc7464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 17:04:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4138e3d5c1af41c68f8ae40973ad7587\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36c81df9c1254991860744b3284eaabd : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56c4812d89724c4c886db68a9abaaaf3 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7f4168c3d75405f9bafb7f556a88297 : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8cd7f4a9121f44f0962e4bc327bbb6de : SUCCESS in 20m 38s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a36f5959f7c346729eb7caa2f4546171 : SUCCESS in 12m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/df065d4371f946b8b1f30a7e0636db77 : SUCCESS in 1h 03m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bccf07c459a24f53975a9313a6cd7a0d : SUCCESS in 17m 39s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b28940525f6f43699369ddb5b55f2584 : SUCCESS in 2h 38m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a853a4a023fb46e591d102daeeb6886a : SUCCESS in 1h 29m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3124a1441fed42359b6721556188b7b0 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d1a19397b9c3592d62c11b4a08bd42c3cf663b7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 17:04:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a8661ab5b5f815c3072083019606b7adec4edac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 17:05:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5287bd46f76f494fa3ac24d30d386134\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3e79cee7ddf439b9ba5946e2adf2ebb : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cce62f65ae7a9eaa9c4c081946874754892c944","revisions":{"1cce62f65ae7a9eaa9c4c081946874754892c944":{"kind":"REWORK","_number":1,"created":"2023-03-31 04:26:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/879122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/22/879122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/879122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/879122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/879122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/22/879122/1"}}},"commit":{"parents":[{"commit":"41ba79c531a383d5037c5373b4bf93203c888386","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/41ba79c531a383d5037c5373b4bf93203c888386"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-06 04:40:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-31 04:26:10.000000000","tz":540},"subject":"Expose executor_thread_pool_size","message":"Expose executor_thread_pool_size\n\nThis option has been supported by puppet-oslo but has not been\nconfigurable.\n\nConflicts:\n\tmanifests/init.pp\n\n(yoga to xena)\nConflicts:\n\tspec/classes/heat_init_spec.rb\n\nChange-Id: If4ea3913c5ec9c3719d57b8b682a87816cbaabf7\n(cherry picked from commit ee09225ba9ea1d68f79dbadd850a878c671c685b)\n(cherry picked from commit 23eb2c50710d37c7d15c390cb013cb1e24692297)\n(cherry picked from commit de3d03502f976c962392e37327bd7b8bd4873463)\n(cherry picked from commit 9bf5e694bfaeeef88f2055fe4c9d52ac4fe191e5)\n(cherry picked from commit 3b41bb0443701e47b0beee648c34c7d7f740675d)\n(cherry picked from commit 059b2e7204fadf69341c69116a79551e4b980e03)\n(cherry picked from commit 5fed3d85fe0928d93f88c3ef1d5016968e667555)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1cce62f65ae7a9eaa9c4c081946874754892c944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1cce62f65ae7a9eaa9c4c081946874754892c944"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
