)]}'
{"id":"openstack%2Fpuppet-heat~886722","triplet_id":"openstack%2Fpuppet-heat~master~Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a","project":"openstack/puppet-heat","branch":"master","topic":"cron-ensure","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-28 15:42:58.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2023-06-28 14:29:49.000000000","reason":"removed on reply"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-06-27 10:42:20.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a","subject":"Validate status of cron resource","status":"MERGED","created":"2023-06-22 07:22:51.000000000","updated":"2023-06-28 15:42:58.000000000","submitted":"2023-06-28 15:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886722-cron-ensure","meta_rev_id":"383c0f216eb51fd6cf148f6a64794f314774f2fb","_number":886722,"virtual_id_number":886722,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 15:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-06-27 10:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-06-28 14:29:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-06-28 14:29:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 08:35:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 09:31:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-06-22 14:33:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-06-26 11:31:19.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"436553e10f76ad776d7b2a877c748d3b4d7e8912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-22 07:22:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c28e3ba760b11d7604d57db90915974933e403e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 08:35:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fed5d609e8d841f49fc17bd077872a91\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3af715811c9412ea48135e961cba13d : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4957bc24ad90432ba5bac994f41d1f47 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a46ceb55ce4496eb9f65989340905d7 : SUCCESS in 5m 16s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf4f4f721b0b4f15adecb71b4d82a20c : FAILURE in 10m 49s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b85c2b9c28334fdd849fe89373db0534 : FAILURE in 8m 20s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bea9f3f05974c07b5becca31567075e : FAILURE in 12m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d89c64acf2b41e1b76db06950a41889 : FAILURE in 9m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9192008f37264e35bc0af397ff71d0e1 : SUCCESS in 51m 01s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/115d11fb0fdb45a8bb43a770803be52f : SUCCESS in 1h 04m 15s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18232f3285694510a54234f6bc87ae81 : SUCCESS in 51m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2436a88e2b74af5a5979e3f71286efa : SUCCESS in 1h 11m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4dd48d35e7449e4ab16a035aa59f7c3 : SUCCESS in 22m 27s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/503399f7fb544a06a899f313cd096301 : SUCCESS in 15m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf4334bc38ca44e092f899a1d752eed8 : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22410181c8dbebede785c6febc36ed664afca749","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-22 09:29:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"745a248e0c2423c829001ba5f309a588a30e8baf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-22 09:31:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdba8eb810a8d1ecd3c8581b8fa0b4a25b636fbb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-22 09:31:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a62fef1817da7705c12f2af3079021ef449c77f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-22 09:50:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f6ae5a2291147654af7a9273fcb9d604d1377e9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:32:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a9a431ddec45fbb737b1f5510a8c49\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c51c9d0ec66e4b349046b8eb6b5eebf1 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8972cb0e5c1d407b8ac425af8b7b5b00 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/513a3392f71f448788ef70898ff34593 : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/bca851406dd54f2484de237b3d9b7294 : SUCCESS in 11m 25s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d67186676994073afcdf5ae627bd1eb : SUCCESS in 11m 36s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/359bacc85f394b03a54a42b93e9afdd6 : FAILURE in 13m 50s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ee2afc625b0448a852de1a16d85aa27 : SUCCESS in 7m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f36f310b19742708b48db7ea3cb8ae5 : SUCCESS in 56m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a32613d90d734d31b5a17c35044ced69 : SUCCESS in 1h 02m 30s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57bd66397c8544bda47b42870fbbd2d7 : SUCCESS in 1h 29m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d8d4ba6d8c3433aada2b3fa754a404e : SUCCESS in 1h 22m 23s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5f8e1359f1b483996d2d0e64bf89805 : SUCCESS in 25m 04s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bff8479cebe428e9a8a55f9129b4876 : SUCCESS in 21m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3918d25bd08e440fb0a67e09b6908d03 : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a91234f5c74d67aeab4e99a08ee431e27d5ea64","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-22 14:33:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1debbbf36744e420a2531270099e54d092c11ebb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-26 11:31:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed7127ad216f179501df75d360db0ec4ba3a9dce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 13:08:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2bacc2b8caba9aea6bb7359e93361c7766c03db1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-26 13:09:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea7db89dcfc9a01242197a33768ac4d19566a39d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 15:51:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d6ad83b44d488191a2c08f02da9e1a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d756b211bf384878b4135375fd1a475d : SUCCESS in 2m 56s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c489224f40344fcda02c8327a934e980 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b4b06ff9cc44552a58689f606fabc11 : SUCCESS in 5m 43s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ead17a6f23b4ee7a2d70f56eb7a48b2 : SUCCESS in 11m 41s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccdb6eb748d54d4aa0cf8266cee27b8f : SUCCESS in 8m 53s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e76f54592bd4e27a95e04861a13476c : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e58197b71d4a088fe99a4e3cdc1b98 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5751a0fa8524f7f9b624926c8eb434c : SUCCESS in 56m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce9c1724e2ec44e19e7e961f67068186 : SUCCESS in 1h 00m 31s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce03832f7bbf420688795a4004ab4c33 : SUCCESS in 53m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/190d27b93e4c4172b005e31748db8e02 : SUCCESS in 1h 19m 24s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cad708cc5e824928ab5098401f735949 : SUCCESS in 23m 56s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1e4a50d3dc14acc88d49b61389078c6 : SUCCESS in 36m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78a80aa03abc47d59b7e938ee2a7733c : SUCCESS in 20m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"512e094c6741d276c661a6ad9947e6cfa399923b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-27 10:42:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f64b66f642ec4ec810e7cfa7153bb6b4826b2a74","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-06-28 14:29:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eab464740658e3eb7280f364c238957d15596562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:30:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3bc7d1d0234de4e3b397b5df2e45572e16a4150f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 15:42:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ebd1de14f748b79e81c5295093f72f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54d8c1ea871241968a2c475ac19e5674 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/537901b3db084df2932af4f3d5bc8bc7 : SUCCESS in 2m 43s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c5c1953f9e847f6918f3700e71d4e73 : SUCCESS in 41m 55s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a56aa20eec7143e3b3cac9ed73d31874 : SUCCESS in 7m 55s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb0e9f5ada0540b89a61cdaebe263178 : SUCCESS in 11m 09s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b0509d5e0454765bb35d01e62672c54 : SUCCESS in 1h 06m 32s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/91c64a9f081044719185ca54105b1293 : SUCCESS in 1h 06m 13s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/701cdeaff9024a4186dc86c79e6313f1 : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":4},{"id":"383c0f216eb51fd6cf148f6a64794f314774f2fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 15:42:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c62209b8e7439431834dacd1f2a0d58599ecf623","revisions":{"2ef80a0a13ece281a18158deabc7e3ee0c155312":{"kind":"REWORK","_number":1,"created":"2023-06-22 07:22:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/886722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/22/886722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/1"}}},"commit":{"parents":[{"commit":"7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-22 07:22:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-22 07:22:09.000000000","tz":540},"subject":"Validate status of cron resource","message":"Validate status of cron resource\n\nThis makes sure that the supported value is used for the parameters\nwhich determine the ensure property of a cron resource type.\n\nChange-Id: Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/2ef80a0a13ece281a18158deabc7e3ee0c155312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/2ef80a0a13ece281a18158deabc7e3ee0c155312"}]},"branch":"refs/heads/master"},"4f648510f51571b7914fca1f8e3541f63f3d0bb2":{"kind":"REWORK","_number":2,"created":"2023-06-22 09:29:50.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/22/886722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/22/886722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/2"}}},"commit":{"parents":[{"commit":"7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-22 07:22:09.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-06-22 09:29:49.000000000","tz":120},"subject":"Validate status of cron resource","message":"Validate status of cron resource\n\nThis makes sure that the supported value is used for the parameters\nwhich determine the ensure property of a cron resource type.\n\nChange-Id: Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4f648510f51571b7914fca1f8e3541f63f3d0bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4f648510f51571b7914fca1f8e3541f63f3d0bb2"}]},"branch":"refs/heads/master"},"402d63903e71211ceefb08f5795ac7a7120a0f8f":{"kind":"REWORK","_number":3,"created":"2023-06-22 09:31:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/22/886722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/22/886722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/3"}}},"commit":{"parents":[{"commit":"7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-22 07:22:09.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-06-22 09:31:05.000000000","tz":120},"subject":"Validate status of cron resource","message":"Validate status of cron resource\n\nThis makes sure that the supported value is used for the parameters\nwhich determine the ensure property of a cron resource type.\n\nChange-Id: Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/402d63903e71211ceefb08f5795ac7a7120a0f8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/402d63903e71211ceefb08f5795ac7a7120a0f8f"}]},"branch":"refs/heads/master"},"c62209b8e7439431834dacd1f2a0d58599ecf623":{"kind":"REWORK","_number":4,"created":"2023-06-26 13:08:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/886722/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/22/886722/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/22/886722/4"}}},"commit":{"parents":[{"commit":"7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7","subject":"Merge \"Remove support for Puppet 6\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7d927b9a02244e5231dfc46a9b7a933b2ac8b4b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-22 07:22:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-26 13:08:16.000000000","tz":540},"subject":"Validate status of cron resource","message":"Validate status of cron resource\n\nThis makes sure that the supported value is used for the parameters\nwhich determine the ensure property of a cron resource type.\n\nThis also re-implements the validation of age_type parameter using\nthe native data type.\n\nChange-Id: Ie3d9c4d3dac3484051be3b9418e833454c0fbd2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c62209b8e7439431834dacd1f2a0d58599ecf623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/c62209b8e7439431834dacd1f2a0d58599ecf623"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
