)]}'
{"id":"openstack%2Fpuppet-heat~956331","triplet_id":"openstack%2Fpuppet-heat~master~I64c364baa5549fbdccb280083bbc7d607e2d8407","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-13 15:11:57.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2025-10-13 14:09:45.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I64c364baa5549fbdccb280083bbc7d607e2d8407","subject":"Add support for [DEFAULT] region_name_for_shared_services","status":"MERGED","created":"2025-08-01 05:36:27.000000000","updated":"2025-10-13 15:12:43.000000000","submitted":"2025-10-13 15:11:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956331","meta_rev_id":"5021a98cdd6d8376c238a4967c8ec9f2bc8cbd0c","_number":956331,"virtual_id_number":956331,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-13 15:11:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-10-13 14:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-10-13 11:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2025-10-13 14:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 06:32:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 18:24:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-10-13 11:52:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3ff64a709485dca963cb5fb3e29559c514c1d0e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-01 05:36:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c25e8d43989b82a41b886e0a99aa52e4295c836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 06:32:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eab1895465d548f9a2f71c1dc84db4de\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ce7606235a843a5ba0259f37a46e12c : SUCCESS in 2m 17s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96669db2ba0d4cffb20a0ec745491919 : SUCCESS in 1m 51s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fccf9a2e889e412c9fd6f26db05df3f1 : SUCCESS in 3m 05s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/85077edcfbb94a67974a2a87502bef23 : FAILURE in 6m 58s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6e6541196aa4e55a8b2508afd045fbb : FAILURE in 4m 03s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d85884a15474bf98480430f7e10e3fc : FAILURE in 2m 43s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d8e5bfae19c4a989471db6b395541f1 : SUCCESS in 37m 34s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e9eb3a45144b9bbfbe10b607e629dc : SUCCESS in 48m 35s\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/716a4c8b39fb4bcab5c56cf1e5f39420 : FAILURE in 29m 45s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8cb912350fe4d349ab147668597af46 : SUCCESS in 29m 52s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8497ddc3ab4f461da47f033289b70ad9 : SUCCESS in 8m 24s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/340c251b324f459f8f1adeb1852764cb : SUCCESS in 11m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbfcb8e42452473f8c4b7c19a5eb863c : SUCCESS in 7m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b64d76482de4ae486e0fc283a03c762 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d84989b871e605a1dba183e41ba8fabee94b326a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-06 14:20:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3702e7eb749c6e01515a4a896cb22136463ffe41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 15:32:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44528a9c528f4a8d87fad29b16d50daf\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/036c3bfde50140d9bfd61b089c465073 : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26a587ce4b0c4ea78269d858b5b11f64 : SUCCESS in 1m 33s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d86332307a54de4a7fa758bf7619a1d : SUCCESS in 2m 34s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/26dcfc2f04a84b3387e0879fd6007a89 : SUCCESS in 9m 17s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/822aac2fe9414ac99179dc93e3752831 : SUCCESS in 5m 05s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18ded55420af43ffba2974d221adb6dc : SUCCESS in 5m 04s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/039c04b432224b59a6d6c9b0ddf41739 : SUCCESS in 44m 48s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/39e1f02a1c0e42eaa71d11c7460a5e4c : SUCCESS in 1h 04m 44s\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7daf5d8a467649e7bfa058bf912992c5 : SUCCESS in 48m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed687980a6b44da5a00cb95a99c620e9 : SUCCESS in 54m 37s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/286a4c9922aa462b9b2c7fe67fa622c4 : SUCCESS in 20m 08s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a237949fa5e8429a942ecb0da35ba26a : SUCCESS in 14m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/956caa88be904fb788f3f60d6841f0c9 : SUCCESS in 13m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cae2b4b913041b582b8ca3897395d87 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5db37ae3fea35cb9deab71e8dd0644c9e985f7cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-06 18:24:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea1af22c691f9c9e4dc13d4c33f501055ebbae3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-13 10:18:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"93ed926313ac2119eb745e88fc2dc60d5e40de37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 10:37:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e5027053aa5495ca06257e1ee554916\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a7f11f06beb4ad5b98158edc0395da0 : FAILURE in 3m 41s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b5048a115f41c3b84af7d1891b3997 : SUCCESS in 2m 53s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/ed40592579ad467eb334e70866ea964f : FAILURE in 7m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/400a7010eca44153ae6d28035b3bc873 : FAILURE in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e9fb5b9f43d4fbab2443b9e74d3085e : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e78d877915c44bf28ba684211f0ae8d8 : FAILURE in 4m 57s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61633cc718a34c36a716f9c5ca0991d4 : FAILURE in 9m 30s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/811396842b61401eb431af6c60a0719c : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f400b85cde7f41f1aaaba6e3c9777a1e : FAILURE in 3m 09s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c29aa70bbe4f4249bad4fdd76ceb8b02 : FAILURE in 11m 43s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb22aabec9d44d55b26d3108ac587a89 : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/02b4fd6ed8c74e958a32e67d1566b2b9 : FAILURE in 7m 35s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc5db8ad0de44052912a4be49aa3fc80 : FAILURE in 5m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f0ccd65d5e2407eb695794e30f7e35a : FAILURE in 2m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3600a5e2e693485b8517be84636b3f76 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9ea4698dec73e5fef82c17b591a35d39f5400ec2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-13 11:06:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be7ad72c813ce073d31c7ea66cd854d90f77bc37","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-13 11:52:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e532c0e7a06440ae62c93ba4ab50e8c586d3c156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 12:33:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c3e107e16df40bd82bf218a27169454\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca82bd2f77934be5bb9733fb2bf23089 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8320f319ed744508b7403d7f5609e340 : SUCCESS in 2m 38s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/24531fde1623430180a52058ddec10ec : SUCCESS in 4m 00s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f4a63921c354a5eb4a75b7129175007 : SUCCESS in 5m 24s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed1dca379b79448c977cfac9018e3423 : SUCCESS in 3m 09s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/033d0157ce864011babba4cfa80d154c : SUCCESS in 43m 25s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a7f16f5a591490f8f28e86871d563f5 : SUCCESS in 1h 21m 46s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32a7854c54574b579b4373d88bcafc27 : SUCCESS in 1h 06m 45s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83b84541ec934e9bb7212c4b1e177471 : SUCCESS in 48m 45s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61424050a11b43f2bd6db77899327b6b : SUCCESS in 29m 16s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54fafac6dd5644e1b29456b59d2ab6c1 : SUCCESS in 58m 13s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b15660bfa22b439eafba09cfe9c8163a : SUCCESS in 14m 09s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/929366fd24094f39a52c3691a1c3c407 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57cee3a3d03c4fe28ade9c432895dcfb : SUCCESS in 10m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aab7e2725ba40579f0a83c15b246624 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":4},{"id":"5fc9016c965bdebb566d58f48149791a306a8fb4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-10-13 14:09:45.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a0aece1c2b19a1015cafd9a138e1082e28047bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 14:10:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bf611bc99c15094b2ac5b092f8140fee9d71bb78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 15:11:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3909c866fa048dea71bd6989690b3e9\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4980f199e6b047138f2f14295c63f129 : SUCCESS in 2m 29s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/869000c109364c7c94f67bbf7fd1352b : SUCCESS in 2m 34s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c982eed533d1473b9d40186f4334b60f : SUCCESS in 4m 45s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d097e65e4c204a59a6bc5a5feb5df4d7 : SUCCESS in 6m 09s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a15bcd94ebd46b78f316262484142a5 : SUCCESS in 53m 59s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d87368ab8be4f22905cfe59482b4dfe : SUCCESS in 55m 26s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cb245b16ce44637bb3694487fcdc2b3 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47dcaae6d3a64bd1baa7034db35ddb34 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":4},{"id":"15f54013925f1cb7e94b8bf628a53b098cb2cd9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 15:11:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5021a98cdd6d8376c238a4967c8ec9f2bc8cbd0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 15:12:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b4ef7167ae04b3a975abcfdc911fc99\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee8368a190094be684046c74ae8851f2 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3ab5c581b9ec5d93fe501129a116084f7fa8ab94","revisions":{"db060d5c7ed68a9aba99a0f1ef7a7cf98c4c15c5":{"kind":"REWORK","_number":1,"created":"2025-08-01 05:36:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/956331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/31/956331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/1"}}},"commit":{"parents":[{"commit":"bcaa08fa4c4d38a427c8fffea0cb911debaebf33","subject":"Fix outdated parameter description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/bcaa08fa4c4d38a427c8fffea0cb911debaebf33"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 05:33:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 05:36:11.000000000","tz":540},"subject":"Add support for [DEFAULT] region_name_for_shared_services","message":"Add support for [DEFAULT] region_name_for_shared_services\n\n... to customize region used for shared services.\n\nAlso add support for [DEFAULT] shared_services_types as well, because\nthis option is tightly related.\n\nChange-Id: I64c364baa5549fbdccb280083bbc7d607e2d8407\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/db060d5c7ed68a9aba99a0f1ef7a7cf98c4c15c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/db060d5c7ed68a9aba99a0f1ef7a7cf98c4c15c5"}]},"branch":"refs/heads/master"},"cef38ea786e4cdb80ea1b2f0188c1f5ef4c5d3af":{"kind":"REWORK","_number":2,"created":"2025-08-06 14:20:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/956331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/31/956331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/2"}}},"commit":{"parents":[{"commit":"bcaa08fa4c4d38a427c8fffea0cb911debaebf33","subject":"Fix outdated parameter description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/bcaa08fa4c4d38a427c8fffea0cb911debaebf33"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 05:33:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-06 14:20:21.000000000","tz":540},"subject":"Add support for [DEFAULT] region_name_for_shared_services","message":"Add support for [DEFAULT] region_name_for_shared_services\n\n... to customize region used for shared services.\n\nAlso add support for [DEFAULT] shared_services_types as well, because\nthis option is tightly related.\n\nChange-Id: I64c364baa5549fbdccb280083bbc7d607e2d8407\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/cef38ea786e4cdb80ea1b2f0188c1f5ef4c5d3af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/cef38ea786e4cdb80ea1b2f0188c1f5ef4c5d3af"}]},"branch":"refs/heads/master"},"30a793c71a1f2f58b26af3c39abf13e10f2a0980":{"kind":"REWORK","_number":3,"created":"2025-10-13 10:18:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/956331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/31/956331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/3"}}},"commit":{"parents":[{"commit":"1fd20bd4de8ab3e11daf5bc5f3b247f927e8cbfa","subject":"Prepare 2025.2 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1fd20bd4de8ab3e11daf5bc5f3b247f927e8cbfa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 05:33:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-13 10:17:52.000000000","tz":540},"subject":"Add support for [DEFAULT] region_name_for_shared_services","message":"Add support for [DEFAULT] region_name_for_shared_services\n\n... to customize region used for shared services.\n\nAlso add support for [DEFAULT] shared_services_types as well, because\nthis option is tightly related.\n\nChange-Id: I64c364baa5549fbdccb280083bbc7d607e2d8407\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/30a793c71a1f2f58b26af3c39abf13e10f2a0980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/30a793c71a1f2f58b26af3c39abf13e10f2a0980"}]},"branch":"refs/heads/master"},"3ab5c581b9ec5d93fe501129a116084f7fa8ab94":{"kind":"REWORK","_number":4,"created":"2025-10-13 11:06:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/956331/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/31/956331/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/31/956331/4"}}},"commit":{"parents":[{"commit":"1fd20bd4de8ab3e11daf5bc5f3b247f927e8cbfa","subject":"Prepare 2025.2 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/1fd20bd4de8ab3e11daf5bc5f3b247f927e8cbfa"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-01 05:33:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-13 11:06:43.000000000","tz":540},"subject":"Add support for [DEFAULT] region_name_for_shared_services","message":"Add support for [DEFAULT] region_name_for_shared_services\n\n... to customize region used for shared services.\n\nAlso add support for [DEFAULT] shared_services_types as well, because\nthis option is tightly related.\n\nChange-Id: I64c364baa5549fbdccb280083bbc7d607e2d8407\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3ab5c581b9ec5d93fe501129a116084f7fa8ab94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/3ab5c581b9ec5d93fe501129a116084f7fa8ab94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
