)]}'
{"id":"openstack%2Fpuppet-heat~962065","triplet_id":"openstack%2Fpuppet-heat~master~I9f44215d2806653aec84076414f1ae20d8b7c1ec","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-25 17:24:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9f44215d2806653aec84076414f1ae20d8b7c1ec","subject":"Validate ensure parameter for package resources","status":"MERGED","created":"2025-09-23 08:34:22.000000000","updated":"2025-09-25 17:24:45.000000000","submitted":"2025-09-25 17:24:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962065","meta_rev_id":"590596a7a5b570d3fed5ced6c5cbe0311d105d7b","_number":962065,"virtual_id_number":962065,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-25 17:24:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-09-23 12:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-09-25 16:10:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-25 16:10:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-23 09:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-23 12:24:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-25 16:10:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e89fe6611b0ef0c82f586076aa777ed27175218c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-23 08:34:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cb3212643b7f44f339b49fe9215bb2aa46532b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-23 08:40:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d3fd4601a434192746d6be9314585ea8867e065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-23 08:42:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a243a660093c90747d2e1b0e69aa30f358b2fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 09:51:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec3fd3985884f638bbfab9e9e2af74a\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/574bc7c0673e4559951a4e9a9df2c314 : SUCCESS in 3m 00s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98195b1323f84fa8a1669ed31c3fde47 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c21fd34b46eb47999dbf64bf17839ca6 : SUCCESS in 1m 50s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/6be395dd044d4704b5decbf97d922b33 : SUCCESS in 6m 11s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3072318de9ed41e4bb589f9fd9a53088 : SUCCESS in 5m 53s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb68169e33d0405a99fb67c012616c9c : SUCCESS in 6m 10s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eed2f00bf4e2472ab37e5027087f1a20 : SUCCESS in 1h 01m 59s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07b00e7d98594472a35b1498aa07164c : SUCCESS in 1h 00m 59s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f718f23faebb49478f2785e25473de10 : SUCCESS in 53m 42s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9315262b4c5441c1bcefa32fecdc63a4 : FAILURE in 29m 42s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8d772bfbd8346d29481852522eaac92 : SUCCESS in 44m 24s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f9873ca82f94c24ac9e65892ef6acc0 : SUCCESS in 59m 06s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/091eb053318e44bbb29271a260664a53 : SUCCESS in 14m 48s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/230b785f2773410ab164e6b1c8954d33 : SUCCESS in 10m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a6f6c22f5a84e52a72daeed7f9e408d : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"545d3983e1a43c96bc4228b09abb65d8d90c2e11","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-23 12:24:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c71ce89ef006f0a49ebdb15620bb566a0313e749","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-25 16:10:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cee0ef77692d21db747301b972357f70a35b6f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 16:11:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4e139b6f2b56b3ca5870a1f3adfb6ad90a63f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 17:24:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceadbbf2b6a94f22bd7f3a07e988d286\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1ff5584844dc4cd6a2a4bd7a3e2a86fa : SUCCESS in 1m 23s\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1654f5fa44b64986a80873a3632b27bc : SUCCESS in 1m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40a3b080504d4af38a3a7f1c9732551d : SUCCESS in 2m 59s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/82b3b2998198470ca499ee3dad57ddee : SUCCESS in 3m 53s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63446c1aadd944d08120100c26d3d65a : SUCCESS in 3m 22s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/787fce9035d04aa89bcccdf8a73714fb : SUCCESS in 1h 08m 48s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/22db23e01bd640efb46c592d2beb9d27 : SUCCESS in 56m 14s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/425db3e5ddce4c08ba4bb1c0e61b0144 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":3},{"id":"590596a7a5b570d3fed5ced6c5cbe0311d105d7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 17:24:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84eb74d6e4b3753f92a0906f969dc0bafb2c1e96","revisions":{"181273dafeb4ffbafd5df53aea2834a47fd7bdca":{"kind":"REWORK","_number":1,"created":"2025-09-23 08:34:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/962065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/65/962065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/1"}}},"commit":{"parents":[{"commit":"213c858298228f5dcac7ba18139e124ad7cd7f54","subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/213c858298228f5dcac7ba18139e124ad7cd7f54"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:34:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:34:13.000000000","tz":540},"subject":"Validate ensure parameter for package resources","message":"Validate ensure parameter for package resources\n\nThe minimum version of puppetlabs-stdlib has been bumped globally, so\nnow we can use the common type definition.\n\nChange-Id: I9f44215d2806653aec84076414f1ae20d8b7c1ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/181273dafeb4ffbafd5df53aea2834a47fd7bdca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/181273dafeb4ffbafd5df53aea2834a47fd7bdca"}]},"branch":"refs/heads/master"},"e0e5b4b7909134d26d441d3dc92e657abae09b00":{"kind":"REWORK","_number":2,"created":"2025-09-23 08:40:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/962065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/65/962065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/2"}}},"commit":{"parents":[{"commit":"213c858298228f5dcac7ba18139e124ad7cd7f54","subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/213c858298228f5dcac7ba18139e124ad7cd7f54"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:34:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:40:31.000000000","tz":540},"subject":"Validate ensure parameter for package resources","message":"Validate ensure parameter for package resources\n\nThe minimum version of puppetlabs-stdlib has been bumped globally, so\nnow we can use the common type definition.\n\nChange-Id: I9f44215d2806653aec84076414f1ae20d8b7c1ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e0e5b4b7909134d26d441d3dc92e657abae09b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e0e5b4b7909134d26d441d3dc92e657abae09b00"}]},"branch":"refs/heads/master"},"84eb74d6e4b3753f92a0906f969dc0bafb2c1e96":{"kind":"REWORK","_number":3,"created":"2025-09-23 08:42:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/962065/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/65/962065/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/65/962065/3"}}},"commit":{"parents":[{"commit":"213c858298228f5dcac7ba18139e124ad7cd7f54","subject":"Remove support for [oslo_messaging_rabbit] heartbeat_in_pthread","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/213c858298228f5dcac7ba18139e124ad7cd7f54"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:34:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-23 08:42:19.000000000","tz":540},"subject":"Validate ensure parameter for package resources","message":"Validate ensure parameter for package resources\n\nThe minimum version of puppetlabs-stdlib has been bumped globally, so\nnow we can use the common type definition.\n\nChange-Id: I9f44215d2806653aec84076414f1ae20d8b7c1ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/84eb74d6e4b3753f92a0906f969dc0bafb2c1e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/84eb74d6e4b3753f92a0906f969dc0bafb2c1e96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
