)]}'
{"id":"openstack%2Fpuppet-heat~962355","triplet_id":"openstack%2Fpuppet-heat~master~I6ef226cf07eae253203ea8ae4412d9530075cf3b","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-29 15:30:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ef226cf07eae253203ea8ae4412d9530075cf3b","subject":"Refactor api service management","status":"MERGED","created":"2025-09-26 14:07:36.000000000","updated":"2025-09-29 15:30:39.000000000","submitted":"2025-09-29 15:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"962355","meta_rev_id":"de0a37dddd95338e973a765bcf5ca71e2a80c5d2","_number":962355,"virtual_id_number":962355,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-29 15:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-09-29 06:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-09-29 14:19:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-09-29 14:19:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-26 16:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 06:19:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-29 14:19:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f3cd79331122bd7895006dd0959f3ab19f2cc260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f437bc9d6dd732a479a28091cc4ca0fbd02cb916","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:13:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b688eb296d159cf92fd375b2e14e8ff611fbdc1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:29:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3840bca0cc87410a86f4020590bf924706aac64e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 15:34:06.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"27e50afa6f2e67c214b583d5089b3297327f59da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 16:50:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/412fcd179e264f25bc66203d543413aa\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/049fd4e5ec36498b8402bdd7d67bd793 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a67c77c990e34de6a86487b5825625cc : SUCCESS in 3m 21s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c75b49ee1a94438a8a6c8f3d147bff1a : SUCCESS in 7m 15s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42f3ff8da84f438c8d1c3bdceaef8281 : SUCCESS in 6m 17s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c669930b0248408ab937d774947a2711 : SUCCESS in 6m 25s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7850697f746849ceb0563c706af3fd0d : SUCCESS in 28m 23s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/20d0311cdfde4b14878269939046b0d6 : SUCCESS in 1h 10m 57s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/edbb61f948c14c20a7295106260d60f7 : SUCCESS in 1h 08m 33s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0555db4165c242f183bce131f6aa6a27 : SUCCESS in 23m 13s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6278fed2c024103afe9db986795ea5a : SUCCESS in 59m 11s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb51e09ce1ed4584a33d2352c19a1589 : SUCCESS in 48m 47s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4565e59f2774455e868793c82ff6a490 : SUCCESS in 20m 53s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/338862a7095e40328a4bbde18a924438 : SUCCESS in 12m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23f7e120de8c469cb9b67715491043b5 : SUCCESS in 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9703c955511173b8b233122024c23a49baaffb16","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-29 06:19:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9dc6495e6234462ddcb4e41340a94709f9783dce","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-29 14:19:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b0a4fd5743828063d1782b4509c45f23edb9c7bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 14:20:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"12d6c665d67d6b0c3977e6014592fdcaec2272d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:30:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e21cf46c074f16a1a6b98d41999e9a\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/317eb1f3401946c1a5ff2e3a2dda2f69 : SUCCESS in 2m 12s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b070659c106e4a0c9af7efa92a92aa03 : SUCCESS in 2m 49s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f418e806ad05463f9ee81c8aa040cfb9 : SUCCESS in 4m 19s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdf417f0eab64bffafa36c0825b7c020 : SUCCESS in 6m 05s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc744f41982f4b199146c162b4d1e842 : SUCCESS in 1h 05m 06s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c16e2372fc854a2ab03c0c721ab9da97 : SUCCESS in 55m 43s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e1e4f165f8649148ade93fd7d8909ab : SUCCESS in 19m 18s","accounts_in_message":[],"_revision_number":4},{"id":"de0a37dddd95338e973a765bcf5ca71e2a80c5d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:30:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"26d9a4e9d1042546e31cab8515032091b3383dd1","revisions":{"e0305d831aa5365e562455b8a8bf5e73d5401553":{"kind":"REWORK","_number":1,"created":"2025-09-26 14:07:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/962355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/55/962355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/1"}}},"commit":{"parents":[{"commit":"59d7f80f9df7e328e487d335d73dce4b63fc76e0","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/59d7f80f9df7e328e487d335d73dce4b63fc76e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:07:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:07:08.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I6ef226cf07eae253203ea8ae4412d9530075cf3b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e0305d831aa5365e562455b8a8bf5e73d5401553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e0305d831aa5365e562455b8a8bf5e73d5401553"}]},"branch":"refs/heads/master"},"5b61fce92a8c9522cd2e11cc45551b5649f78dbc":{"kind":"REWORK","_number":2,"created":"2025-09-26 14:13:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/962355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/55/962355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/2"}}},"commit":{"parents":[{"commit":"59d7f80f9df7e328e487d335d73dce4b63fc76e0","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/59d7f80f9df7e328e487d335d73dce4b63fc76e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:07:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:13:51.000000000","tz":540},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nChange-Id: I6ef226cf07eae253203ea8ae4412d9530075cf3b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/5b61fce92a8c9522cd2e11cc45551b5649f78dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/5b61fce92a8c9522cd2e11cc45551b5649f78dbc"}]},"branch":"refs/heads/master"},"0634f46416bf0496513e4a45d141b010cf81efb7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-26 15:29:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/962355/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/55/962355/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/3"}}},"commit":{"parents":[{"commit":"59d7f80f9df7e328e487d335d73dce4b63fc76e0","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/59d7f80f9df7e328e487d335d73dce4b63fc76e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:07:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:29:48.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons. This allows us to simplify handling of the provided\nservice name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I6ef226cf07eae253203ea8ae4412d9530075cf3b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0634f46416bf0496513e4a45d141b010cf81efb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0634f46416bf0496513e4a45d141b010cf81efb7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"26d9a4e9d1042546e31cab8515032091b3383dd1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-26 15:34:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/962355/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/55/962355/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/55/962355/4"}}},"commit":{"parents":[{"commit":"59d7f80f9df7e328e487d335d73dce4b63fc76e0","subject":"Merge \"Validate ensure parameter for package resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/59d7f80f9df7e328e487d335d73dce4b63fc76e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 14:07:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-26 15:34:06.000000000","tz":0},"subject":"Refactor api service management","message":"Refactor api service management\n\nDrop validation of service name which is not implemented for\nthe other daemons, to simplify handling of the provided service name.\n\nThis allows us to more easily offload the service name definition to\nhiera data in the near future.\n\nChange-Id: I6ef226cf07eae253203ea8ae4412d9530075cf3b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/26d9a4e9d1042546e31cab8515032091b3383dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/26d9a4e9d1042546e31cab8515032091b3383dd1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
