)]}'
{"id":"openstack%2Fpuppet-heat~968119","triplet_id":"openstack%2Fpuppet-heat~master~Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24","project":"openstack/puppet-heat","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-28 09:54:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24","subject":"Deprecate support for [ec2authtoken] auth_uri","status":"MERGED","created":"2025-11-22 14:32:00.000000000","updated":"2025-11-28 09:55:36.000000000","submitted":"2025-11-28 09:54:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968119","meta_rev_id":"f28fef6b8990eca5df23ee217659af549da48053","_number":968119,"virtual_id_number":968119,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 09:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-11-26 15:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-11-28 08:38:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-11-28 08:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-22 17:44:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 15:03:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-11-28 08:38:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"856f29715eae5739655fbad8f75c417b4c59fe7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 14:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b01c3e852cb5187a82d437a41ea52fc45e2f8355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 14:34:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3a35ea3fdea162a675b89e8da16ebe6d935d9bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 14:55:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a37873272005113ae626ed16872d581abd9ccc57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 16:14:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"dee7122b00338572c53111aefe5439475da9b90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 17:44:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83cc6e70c66d49349a02e8bf7b7a1361\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f61e1f41bcf04fc78fa906e48ce86297 : SUCCESS in 2m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d82c3889a7f452285b5a65736f8d85e : SUCCESS in 2m 31s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/0f75814171cb439db5049d9ecbf83e72 : FAILURE in 4m 27s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29d9757d88c341778a3ddff8cef51b2b : FAILURE in 5m 35s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9767ebf96a64414e9fdfded0c1c1d076 : FAILURE in 6m 29s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a15685706a24d0289fe0cdb621b5a38 : SUCCESS in 59m 49s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a8f359570984fd7b2a2d2702d138fb5 : FAILURE in 1h 25m 51s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ff6839646e84b6a89c6bc0137c95d35 : FAILURE in 1h 09m 41s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6194cc3e6e434c89a0e47ffb2cf44e5d : SUCCESS in 46m 01s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c591b2b2c3364ba0b8ad4e9f351c3705 : SUCCESS in 51m 01s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3d920887a724159bdf65141b7537777 : SUCCESS in 51m 03s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae053d315cb146a99751510df7b658c5 : SUCCESS in 10m 26s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ba9e3cc6a83432c88ada9fb2e402243 : FAILURE in 38m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/351a325ed8144cae9004147f44e000d0 : FAILURE in 38m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ba9a8f5a98f4f57933d915c68a28f37 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8393c5f844e2ab132c11ce9f50eb0e9ef5db5716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-22 17:46:17.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4e184092a7af81cba3b76f7e81359cb348e4c383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 19:27:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51c9772c4d1b4ed09838adf2bd2614a5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d9f5ad2b4504f43bb66e1d6b782d8b1 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93ac3dde534046a5aac7a38aca52c8cc : SUCCESS in 2m 51s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/040f3e8d69d34594ac422381884bc8d4 : FAILURE in 5m 51s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e078aee95b2045cd89b212e7ed9d5f9f : FAILURE in 6m 07s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52106e7d5c6a4423b203e4d203a78941 : FAILURE in 6m 02s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d01a4fb365d4402ca2a7aeaaba6a9a0f : SUCCESS in 1h 00m 09s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7282ae824df47ec94f6cf38745e85c9 : FAILURE in 1h 29m 39s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f85556a918ca4ae1b3533e2ab533a744 : FAILURE in 1h 25m 03s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/317df073f1974b2fab60d46ba69c1200 : SUCCESS in 50m 13s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/178fb55fba9448ad96d0cad7710aa38a : SUCCESS in 53m 43s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bd83393b2fc4db2958381fd05c3a366 : SUCCESS in 57m 13s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9639800b7c440a0a8588abb841e1bd8 : SUCCESS in 20m 30s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/005f320e417849bd9b1e8c92a7e8152a : FAILURE in 39m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b0fd8a730cb436aa70d98994bb57af6 : FAILURE in 38m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d481308443b24d83b2a05bf5d993da64 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":5},{"id":"1ac544cd98ae3bc20f877f7bd31b9249bc8ee5ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 04:08:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4f3c97899f5a5f7fea72f2d572da7f84ec8f818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 05:19:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3708118ccc242888e44eab5a2b38dd2\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0091a073a8f74f30a8facaf124455f86 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e8c98d0a7d46479baa533405018ac8f0 : SUCCESS in 1m 54s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/f2e5d403436e4df0af25ed6792658e87 : FAILURE in 4m 21s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7acc979c5d6425a9fae740a0c9d124f : FAILURE in 5m 48s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80bcedffdfef4a769d38faf8b11fe9cb : FAILURE in 5m 52s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fc5538b3aac458dbeafd51f79b218ec : SUCCESS in 58m 56s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3618bb30dd394b8293ddac31e87b0f1c : SUCCESS in 1h 05m 59s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b2af16cee834ecb8d0fc0f09198893b : SUCCESS in 35m 08s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81a127438d9a44f581235aebf6aaee70 : SUCCESS in 24m 28s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/833dc2ffd63b403f86b10ebd008682d1 : SUCCESS in 57m 50s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/737124719d8648649f96e226c91b3d2c : SUCCESS in 57m 15s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/38b9dd5ae05a48dc93c0f3508939b30a : SUCCESS in 20m 02s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b15eba0d3c8f42048241e2dab5159fff : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5c482aa232e40b48cccb47deb4ea070 : FAILURE in 37m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/443aa9d75bbb4dfd81a752f7e26d97e7 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":5},{"id":"eae1d69479fa5018521f4a41beaa06b96478b286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 05:27:16.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a7fe60b15b834d909a624ba995c4d0614eca94ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 06:36:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42556875e5424be1b8ba6ff388d862b6\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e2ca7484aab40eab360f7dbf65362df : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cf9ebf407274d2197f9edf648014b2b : SUCCESS in 2m 37s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/e6362e0222a648608e7d1e9d3faa82c5 : FAILURE in 4m 22s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6160486318442b7814bebbbb0b704a7 : FAILURE in 5m 49s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f3558a440b744bc95a79ebba2fe5f27 : FAILURE in 3m 41s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec5b5793cb624123b48effcf263b3b39 : SUCCESS in 1h 01m 30s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/18320f611e8a4557806ce68994c03712 : SUCCESS in 52m 22s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac7b682dfc4d4336a7d443aaa86e8d09 : SUCCESS in 55m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb9f100bb31540a589b2eb382d148cf1 : FAILURE in 28m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/66963ef529a944c283e1a192d5e91328 : SUCCESS in 59m 27s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/36cef55e6fa340caaf69c23eec14c3ae : SUCCESS in 45m 23s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ae71c31bcc94edd9597b2d7e1bea5dc : SUCCESS in 9m 45s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/082e76f6f92042b0ac8fef6b681a90aa : FAILURE in 39m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba1f16a810b5478f828dadc80ebdc64d : FAILURE in 39m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8441db69fe64423f9288ec3ca2e16f6d : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":6},{"id":"37c9a6d5dd48d4d80b41d93587eec4c8d59e5963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 10:07:51.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"03a5a2d780a5f005c173dd7d99c28a21bc0bbf80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:23:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c48b8129fb34080b7795b5d26c717b8\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/556b0b4015d14ccd9af86c4d7e784e16 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87eeb3a1175743bab9c68dcc0cdfe269 : SUCCESS in 3m 04s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f9b84f24f8c4de4a62281f695995966 : FAILURE in 5m 20s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a2e9917bd1643e489b2a5de6d935050 : FAILURE in 4m 50s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/811d99e1651c4551a146ec75ef041b91 : FAILURE in 4m 55s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/335f226b514e4f38815d1047bd11e430 : SUCCESS in 1h 00m 50s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/54bfe018f1bd4b84b6a7418c1092b8c9 : SUCCESS in 1h 08m 55s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4f01d0230d2496d9a5c5611e74d4b2c : SUCCESS in 1h 05m 41s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67fdd80c05a54990a3a40440aedf1526 : SUCCESS in 47m 46s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92a3bc89c1e2410fa652952cb73f5e36 : SUCCESS in 1h 03m 19s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/636ed1e3dec246e68be6bd1786e5ef2f : SUCCESS in 1h 00m 14s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/43f53ddcf8f7400393fb0ae05e581648 : SUCCESS in 23m 20s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6dfdb34993d4d1b947825806d73e85b : FAILURE in 38m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/239d6e8c367b40308549ad17ce75fe0d : FAILURE in 39m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c0e0c4974d4b38b00e4f76f9a0b7eb : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":7},{"id":"3a2b8418c19990b34e5737996848544b50540f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-25 13:55:14.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1b9e2749770c6767e8db02381eadc759a7b41dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 15:19:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6cf9c08aa14893988a9bc6f8c4abc3\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69fa540cee32434ab951fdee05356f9a : SUCCESS in 1m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abf428f6d8fe46d08d57a5cf73c7ecba : SUCCESS in 2m 49s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/6bcfd49f45964fdeba449b8dd8dcbd48 : SUCCESS in 5m 22s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b6cea980144cc69a922e0847554761 : SUCCESS in 6m 15s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72478c8306f844adae1f3be59121bd07 : SUCCESS in 3m 16s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6555947d65174f41a75d6b1032224671 : SUCCESS in 1h 03m 12s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9d9702d422b4aeba624cde5b285a006 : SUCCESS in 1h 19m 13s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6a175d69a1b4abd8549f2b57fb2f82f : SUCCESS in 34m 19s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed978c619a5c4c82be613ad7bf3e83d7 : FAILURE in 31m 29s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd71ccd23c6b461d9495322ef345296e : SUCCESS in 1h 00m 26s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2472d29e297049089da72598cc71efd9 : SUCCESS in 32m 13s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8999b72da15c4a49b68b365cc6e54dc5 : SUCCESS in 17m 39s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7472dd0419e4637bc5bd9f248101d32 : FAILURE in 36m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0bf45c8fd9e41ccab31e398082a6342 : FAILURE in 39m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c6bc3929f24d1ba3510a28d7c3a08f : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":8},{"id":"e3ba8ca02324962bc55e449d62ca42a2843d1b46","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-11-26 15:03:24.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1731f16f2ef4dcf938c30117bae222b2d4fdf3ed","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-11-28 08:38:59.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1913b501b5cad41ca805f06e63631d258f8fee5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 08:39:45.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d1513491f25ddd02dae827c2cadbe8bd175f905d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:54:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490de05083f9436fbb68e0db4755091b\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a1cef56a71b4d7d9de73d1e9a59e4a6 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/092a359672174142a2f90dcd01fa818d : SUCCESS in 2m 40s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/cb77c19065fb4d45a4e5859c7313720d : SUCCESS in 6m 14s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/033e85372828485e85cbbee50a1a4f47 : SUCCESS in 5m 42s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b3f84c434e94b4c9b77b53b7deff390 : SUCCESS in 1h 03m 47s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0cdd0579dd9b4ac0ad857f954d427c3d : SUCCESS in 1h 06m 42s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/823bf4f904e14e6395675cccefd91f7b : SUCCESS in 17m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69b568ffa43d44f39a90f98f3f73f3c1 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":8},{"id":"76980924b55fd9e6d0dfbe7ae93fb14324402513","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:54:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"f28fef6b8990eca5df23ee217659af549da48053","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:55:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b35dc17f0147dabd8ffcb3bd901162\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e06e0a5affb478795fe8c4d0f671125 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"caf22939e4578acdbfc76dfd989fd16bf0f337fd","revisions":{"e6b112ac0edd9b6e7074ba31e28ddac96a4d46b8":{"kind":"REWORK","_number":1,"created":"2025-11-22 14:32:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/1"}}},"commit":{"parents":[{"commit":"ecc563cf43328312592a0dbbe14111cf22e86f5b","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/ecc563cf43328312592a0dbbe14111cf22e86f5b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:45.000000000","tz":540},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e6b112ac0edd9b6e7074ba31e28ddac96a4d46b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/e6b112ac0edd9b6e7074ba31e28ddac96a4d46b8"}]},"branch":"refs/heads/master"},"4a1e9b599c06edb133fa775c96ac060447e1ca5d":{"kind":"REWORK","_number":2,"created":"2025-11-22 14:34:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/2"}}},"commit":{"parents":[{"commit":"224b2f6ac25487cc87a7041d28e695aa1a456713","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/224b2f6ac25487cc87a7041d28e695aa1a456713"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:34:14.000000000","tz":540},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4a1e9b599c06edb133fa775c96ac060447e1ca5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4a1e9b599c06edb133fa775c96ac060447e1ca5d"}]},"branch":"refs/heads/master"},"4e9fc82750570df4e05e701bbea48efbf97aedd5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-22 14:55:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/3"}}},"commit":{"parents":[{"commit":"5ab371587dff5c4cf67e6f5741986cef79428096","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/5ab371587dff5c4cf67e6f5741986cef79428096"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:55:22.000000000","tz":540},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4e9fc82750570df4e05e701bbea48efbf97aedd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/4e9fc82750570df4e05e701bbea48efbf97aedd5"}]},"branch":"refs/heads/master"},"8166693732e1b974ec9901f4bbeededc420b362d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-22 16:14:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/4"}}},"commit":{"parents":[{"commit":"48102ff7a3e03a41b8380e9bf7f7b4e89efd8489","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/48102ff7a3e03a41b8380e9bf7f7b4e89efd8489"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 16:14:11.000000000","tz":540},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8166693732e1b974ec9901f4bbeededc420b362d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/8166693732e1b974ec9901f4bbeededc420b362d"}]},"branch":"refs/heads/master"},"11408388832c8ad75fd856e1199f5227af81f4dd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-22 17:46:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/5"}}},"commit":{"parents":[{"commit":"70e69c4e85a81331150ba8095dbfc20a84715195","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/70e69c4e85a81331150ba8095dbfc20a84715195"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 17:46:17.000000000","tz":0},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/11408388832c8ad75fd856e1199f5227af81f4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/11408388832c8ad75fd856e1199f5227af81f4dd"}]},"branch":"refs/heads/master","description":"Rebase"},"267b2dd09cff20b335c96601905458f236b5f383":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-11-25 05:27:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/6"}}},"commit":{"parents":[{"commit":"b2b8e7bd6a0be5e17ce71e31535a948edd270d7f","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/b2b8e7bd6a0be5e17ce71e31535a948edd270d7f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-25 05:27:16.000000000","tz":0},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/267b2dd09cff20b335c96601905458f236b5f383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/267b2dd09cff20b335c96601905458f236b5f383"}]},"branch":"refs/heads/master","description":"Rebase"},"16bb3f6b439d03fa6e8315825ab67be63f3708ab":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-11-25 10:07:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/7"}}},"commit":{"parents":[{"commit":"7d557bf21db320b73beb8e48b7afa07eb71cfd6d","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/7d557bf21db320b73beb8e48b7afa07eb71cfd6d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-25 10:07:51.000000000","tz":0},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/16bb3f6b439d03fa6e8315825ab67be63f3708ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/16bb3f6b439d03fa6e8315825ab67be63f3708ab"}]},"branch":"refs/heads/master","description":"Rebase"},"caf22939e4578acdbfc76dfd989fd16bf0f337fd":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-11-25 13:55:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/968119/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/19/968119/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/19/968119/8"}}},"commit":{"parents":[{"commit":"67e5f5d531793ab35e072718e0e9c915ae8999b1","subject":"Add support for [ec2authtoken] options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/67e5f5d531793ab35e072718e0e9c915ae8999b1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-22 14:31:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-11-25 13:55:14.000000000","tz":0},"subject":"Deprecate support for [ec2authtoken] auth_uri","message":"Deprecate support for [ec2authtoken] auth_uri\n\nThe parameter was deprecated in favor of the new keystoneauth options\nin the [ec2authtoken] section.\n\nChange-Id: Ic4e8ec0e0cd8d4254c061e141090cb8ff0bedc24\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/caf22939e4578acdbfc76dfd989fd16bf0f337fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/caf22939e4578acdbfc76dfd989fd16bf0f337fd"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
