)]}'
{"id":"openstack%2Fpuppet-heat~997699","triplet_id":"openstack%2Fpuppet-heat~master~Ifc5180caed54ac5031219e20a45217104238a3fb","project":"openstack/puppet-heat","branch":"master","hashtags":[],"change_id":"Ifc5180caed54ac5031219e20a45217104238a3fb","subject":"Deprecate support for [clients_vitrage] options","status":"NEW","created":"2026-07-17 09:39:06.000000000","updated":"2026-07-17 11:07:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d02e8b3d153d75e1b50469462ce9907c2029c209","_number":997699,"virtual_id_number":997699,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-17 11:07:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-17 11:07:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2680620cadf9a309ffbdd5a2d612c721bb05326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 09:39:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d02e8b3d153d75e1b50469462ce9907c2029c209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 11:07:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd85d3f0f2148ed8c457b40f172fd85\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05506c03dcb94c63a0eb47587e0fc91c : SUCCESS in 2m 04s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ec3562c4e2a4432bfd9a51e1c9145e4 : SUCCESS in 3m 03s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/d468399db5e44e3995d4c532709ab9df : SUCCESS in 15m 18s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1102fe162d7c4b9f9bc923eccdd6b075 : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ef8759935104bbf9c858611f1e7c096 : SUCCESS in 6m 19s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6b5c1cb06a64928bc3772773c53f594 : SUCCESS in 1h 05m 36s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf8bdc6dd26e4c6a9acb37adcdadef8d : SUCCESS in 49m 30s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db2461d750e04d4c94b13db5b5f119f3 : SUCCESS in 1h 02m 09s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dc8fba2622e4825ab013ecbafea2c94 : FAILURE in 1h 21m 54s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7b947addc7b4225855229228f87a1d2 : SUCCESS in 1h 01m 59s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/644ba0e5113d4c0f8510c0b7932fcb67 : SUCCESS in 52m 57s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/eb641c5edc6144d387cb8b5c117fa701 : SUCCESS in 10m 21s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08693ac028034adda5e2825c78d3186a : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59e1a93f0beb44fe97626b49f2924af7 : SUCCESS in 9m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7429bd70337a403eb3aebb021314787e : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17dea8beaf06fa51fb3ce26fa6f7b1398ecb616c","revisions":{"17dea8beaf06fa51fb3ce26fa6f7b1398ecb616c":{"kind":"REWORK","_number":1,"created":"2026-07-17 09:39:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/997699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-heat","ref":"refs/changes/99/997699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/997699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/997699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-heat refs/changes/99/997699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-heat refs/changes/99/997699/1"}}},"commit":{"parents":[{"commit":"0d0d90e1d80a9b2ec5bda1b69fb4afd84a1ca611","subject":"Remove PostgreSQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/0d0d90e1d80a9b2ec5bda1b69fb4afd84a1ca611"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 09:37:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 09:38:59.000000000","tz":540},"subject":"Deprecate support for [clients_vitrage] options","message":"Deprecate support for [clients_vitrage] options\n\nVitrage resources were deprecated in heat a few cycles ago because\nthe vitrage project was marked inactive, and these are now fully\nremoved because the vitrage project was retired.\n\nChange-Id: Ifc5180caed54ac5031219e20a45217104238a3fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/17dea8beaf06fa51fb3ce26fa6f7b1398ecb616c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-heat/commit/17dea8beaf06fa51fb3ce26fa6f7b1398ecb616c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
