)]}'
{"id":"openstack%2Fpuppet-horizon~656758","triplet_id":"openstack%2Fpuppet-horizon~master~I1aaaef47d4bf29c486ed244cbbe0ada08d84bf87","project":"openstack/puppet-horizon","branch":"master","topic":"os_tempest","hashtags":[],"change_id":"I1aaaef47d4bf29c486ed244cbbe0ada08d84bf87","subject":"Port puppet-horizon-tripleo-standalone to os_tempest","status":"MERGED","created":"2019-05-02 09:56:54.000000000","updated":"2019-05-03 19:08:08.000000000","submitted":"2019-05-03 19:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656758-1556910488774-f36ff123","meta_rev_id":"454c4e24827795470fc4fbd0e825887cd787e148","_number":656758,"virtual_id_number":656758,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-03 19:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-02 11:13:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-05-02 15:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-05-03 17:51:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 11:13:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 15:50:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-03 17:51:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-03 19:08:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab05d5a54a169e1382b6742a93aaa73f59e82b00","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-05-02 09:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d37f33d0997730a2e7bf985b6a0cef59a89cd039","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-02 11:13:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/58/656758/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/7f74125/ : FAILURE in 56m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/58/656758/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/14c2c07/ : SUCCESS in 1h 13m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/58/656758/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a49d0d8/ : SUCCESS in 1h 04m 12s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/58/656758/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/d74664d/ : SUCCESS in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9076891353e7a80103f26bca7869b769c2cb8f8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 11:22:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-lint-ubuntu-bionic/c7e916f/ : SUCCESS in 3m 37s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-syntax-4-ubuntu-bionic/1940283/ : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-syntax-5-ubuntu-bionic/507fa78/ : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-syntax-6-ubuntu-bionic/18db088/ : SUCCESS in 3m 37s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/656758/1/check/puppet-openstack-unit-4.8-centos-7/b85c1e3/ : SUCCESS in 16m 08s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/58/656758/1/check/puppet-openstack-unit-5.5-centos-7/a9fcd73/ : SUCCESS in 14m 33s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/6b67a84/ : SUCCESS in 9m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-unit-latest-ubuntu-bionic/3ca0de7/ : SUCCESS in 8m 31s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/50ec606/ : FAILURE in 26m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/58/656758/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/b2e19f6/ : SUCCESS in 1h 12m 20s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/656758/1/check/puppet-openstack-beaker-centos-7/d90c2bb/ : SUCCESS in 15m 17s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/58/656758/1/check/puppet-openstack-beaker-ubuntu-bionic/a250b9e/ : FAILURE in 9m 53s (non-voting)\n- puppet-horizon-tripleo-standalone http://logs.openstack.org/58/656758/1/check/puppet-horizon-tripleo-standalone/82415bb/ : FAILURE in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e29cf60e6feaa3e64b8e5606eb7e2686db65d70","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-02 15:50:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28f297b8a574bb8a4231f0c43215e104a5a13a02","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-03 17:51:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf4ca064142d4bf8904ba1ad2402b44901505b5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 17:51:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30b7bb1c2e3e2f09abf7be07c179237ddd9bda57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:08:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"221c510995f104fb5f8f580f7b87479d992c244b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 19:08:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-lint-ubuntu-bionic/0e0eae0/ : SUCCESS in 6m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/76fb69c/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/25c0ae0/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/dd194ba/ : SUCCESS in 3m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/58/656758/1/gate/puppet-openstack-unit-4.8-centos-7/d1bd17c/ : SUCCESS in 16m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/58/656758/1/gate/puppet-openstack-unit-5.5-centos-7/f3ab18e/ : SUCCESS in 19m 22s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/d160725/ : SUCCESS in 12m 41s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/58/656758/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/3db39f2/ : SUCCESS in 59m 20s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/58/656758/1/gate/puppet-openstack-beaker-centos-7/be5f860/ : SUCCESS in 20m 29s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/58/656758/1/gate/puppet-openstack-beaker-ubuntu-bionic/4056c3e/ : FAILURE in 9m 17s (non-voting)\n- puppet-horizon-tripleo-standalone http://logs.openstack.org/58/656758/1/gate/puppet-horizon-tripleo-standalone/7697795/ : FAILURE in 1h 08m 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76a3aa2913510e82bb12ca5d62e2009b76bf9a28","revisions":{"76a3aa2913510e82bb12ca5d62e2009b76bf9a28":{"kind":"REWORK","_number":1,"created":"2019-05-02 09:56:54.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/58/656758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/58/656758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/656758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/656758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/656758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/58/656758/1"}}},"commit":{"parents":[{"commit":"daddecad4783b700fc107f8976a54d76cb22c2a8","subject":"Fix dashboard resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/daddecad4783b700fc107f8976a54d76cb22c2a8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2019-05-02 09:56:28.000000000","tz":0},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2019-05-02 09:56:42.000000000","tz":0},"subject":"Port puppet-horizon-tripleo-standalone to os_tempest","message":"Port puppet-horizon-tripleo-standalone to os_tempest\n\nPreviously Tripleo validate-tempest role was used to run tempest\non tripleo jobs. Recently we are moving to os_tempest (an unified\nrole developed by openstack-ansible team).\n\nIt moves the vars to os_tempest.\nBelow is the other var changes:\nrun_tempest: false -\u003e to disable validate-tempest role\nuse_os_tempest: true -\u003e enable os_tempest\ntempest_whitelist -\u003e tempest_test_whitelist\ntempest_plugins -\u003e tempest_services now just the name of the service,\nnot the plugin package name\n\nChange-Id: I1aaaef47d4bf29c486ed244cbbe0ada08d84bf87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/76a3aa2913510e82bb12ca5d62e2009b76bf9a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/76a3aa2913510e82bb12ca5d62e2009b76bf9a28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
