)]}'
{"id":"openstack%2Fpuppet-horizon~658809","triplet_id":"openstack%2Fpuppet-horizon~master~If3e57c7ad0ba72a50fa1b00fe632aa9b60b9e0d6","project":"openstack/puppet-horizon","branch":"master","topic":"remove-lbaas","hashtags":[],"change_id":"If3e57c7ad0ba72a50fa1b00fe632aa9b60b9e0d6","subject":"Remove enable_lb comment","status":"MERGED","created":"2019-05-13 13:04:52.000000000","updated":"2019-05-16 01:53:44.000000000","submitted":"2019-05-16 01:53:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658809-1557971624899-ec554898","meta_rev_id":"d98c2aebe290b0b0c3986466874f881a291b7b53","_number":658809,"virtual_id_number":658809,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-16 01:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-13 15:04:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-16 00:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-15 22:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-16 00:27:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-13 15:04:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-15 22:25:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-16 00:27:58.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-05-16 01:53:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d69a9d0b175ee61972ddee7938399b4ed24b207","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-13 13:04:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ee8705759e3c6cd65c59d3f29a6933d55a8082c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 14:18:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-lint-ubuntu-bionic/0496f25/ : SUCCESS in 3m 43s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-syntax-4-ubuntu-bionic/7cee16c/ : SUCCESS in 6m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-syntax-5-ubuntu-bionic/0a2acae/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-syntax-6-ubuntu-bionic/665cc1c/ : SUCCESS in 6m 22s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/09/658809/1/check/puppet-openstack-unit-4.8-centos-7/6c4b4bf/ : SUCCESS in 16m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/09/658809/1/check/puppet-openstack-unit-5.5-centos-7/49e75b3/ : SUCCESS in 15m 48s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/7a01c12/ : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-unit-latest-ubuntu-bionic/8a6354b/ : SUCCESS in 12m 33s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic/cb5b31c/ : FAILURE in 25m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/09/658809/1/check/puppet-openstack-integration-5-scenario003-tempest-centos-7/7c4ad62/ : SUCCESS in 1h 01m 51s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/09/658809/1/check/puppet-openstack-beaker-centos-7/2b4ee0c/ : SUCCESS in 17m 40s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/09/658809/1/check/puppet-openstack-beaker-ubuntu-bionic/12d4c37/ : FAILURE in 12m 19s (non-voting)\n- puppet-horizon-tripleo-standalone http://logs.openstack.org/09/658809/1/check/puppet-horizon-tripleo-standalone/11ef331/ : FAILURE in 1h 08m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e44e317b510ea09d524cfcc5d33e02c1eeb4c713","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-13 15:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/09/658809/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/3b305f2/ : SUCCESS in 1h 47m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/09/658809/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/1d457a7/ : SUCCESS in 1h 56m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/09/658809/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/1911071/ : SUCCESS in 1h 06m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/09/658809/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/a2e8bf2/ : SUCCESS in 1h 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bcb4f08cfb94f856e39f1f99a9b237f6fd61527","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-15 22:25:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f678bb0c6359dabda711e554e6cbc590612cd24","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 00:27:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"423cbb3e5f76563d5ba71b52a0a44bb9d453ec2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 00:28:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddd321fcc8c9da26fc453a769f7e6bb2bf67d864","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:53:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-lint-ubuntu-bionic/2eb9bc1/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/ea7ebe4/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/7b35116/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/ca0204a/ : SUCCESS in 4m 12s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/09/658809/1/gate/puppet-openstack-unit-4.8-centos-7/c06777a/ : SUCCESS in 18m 07s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/09/658809/1/gate/puppet-openstack-unit-5.5-centos-7/2f2c120/ : SUCCESS in 17m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/a2ab153/ : SUCCESS in 10m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 http://logs.openstack.org/09/658809/1/gate/puppet-openstack-integration-5-scenario003-tempest-centos-7/fd7779c/ : SUCCESS in 1h 06m 29s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/09/658809/1/gate/puppet-openstack-beaker-centos-7/d3663d7/ : SUCCESS in 33m 42s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/09/658809/1/gate/puppet-openstack-beaker-ubuntu-bionic/c2588e9/ : FAILURE in 9m 25s (non-voting)\n- puppet-horizon-tripleo-standalone http://logs.openstack.org/09/658809/1/gate/puppet-horizon-tripleo-standalone/788adae/ : FAILURE in 1h 06m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"777aa566647b0c208eea6b870c60ac205f0f763e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:53:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31c74144c6e9c109ff2edc29f1dbbc1f88284d4e","revisions":{"31c74144c6e9c109ff2edc29f1dbbc1f88284d4e":{"kind":"REWORK","_number":1,"created":"2019-05-13 13:04:52.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/09/658809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/09/658809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/09/658809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/09/658809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/09/658809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/09/658809/1"}}},"commit":{"parents":[{"commit":"76a3aa2913510e82bb12ca5d62e2009b76bf9a28","subject":"Port puppet-horizon-tripleo-standalone to os_tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/76a3aa2913510e82bb12ca5d62e2009b76bf9a28"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-13 13:04:30.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-13 13:04:30.000000000","tz":120},"subject":"Remove enable_lb comment","message":"Remove enable_lb comment\n\nNeutron LBaaS will be removed in commit [1].\n\n[1] https://review.opendev.org/#/c/658801/\n\nChange-Id: If3e57c7ad0ba72a50fa1b00fe632aa9b60b9e0d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/31c74144c6e9c109ff2edc29f1dbbc1f88284d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/31c74144c6e9c109ff2edc29f1dbbc1f88284d4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
