)]}'
{"id":"openstack%2Fpuppet-horizon~703871","triplet_id":"openstack%2Fpuppet-horizon~master~Ic34170530b260b426fd65ac96aa5f494591c2ff1","project":"openstack/puppet-horizon","branch":"master","topic":"bug/1860608","hashtags":[],"change_id":"Ic34170530b260b426fd65ac96aa5f494591c2ff1","subject":"Adding the HTTPONLY cookie header to secure_cookies","status":"MERGED","created":"2020-01-22 21:27:16.000000000","updated":"2020-01-27 18:03:49.000000000","submitted":"2020-01-27 18:03:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703871-1580148229085-5f1f21a0","meta_rev_id":"ca9aee7d2665d5bb7ea7d071eb0e53dc5ed426d7","_number":703871,"virtual_id_number":703871,"owner":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-24 19:40:33.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"date":"2020-01-24 13:33:31.000000000","_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 18:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2020-01-24 21:57:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-01-25 13:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-01-27 16:23:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 19:40:33.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-01-25 13:41:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-01-27 16:23:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-01-27 18:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec31565edfd95f70d6b850ad27dc058634e35248","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-22 21:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8197125ae56a0fb75d416674a6f647ac97e4f4c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 23:00:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/822172050ef84d8a9f0939fa0c2bd1d8 : SUCCESS in 6m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c0bfc156d4440c883b8e0bda91f30c2 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19a8ee0718fa4f5399369ae721dd310c : SUCCESS in 4m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5de95d2c2141457db906787bc2932e01 : SUCCESS in 15m 36s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8704e4a28012411ca93f0566df70b71a : SUCCESS in 13m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97467d031d564ac197f7d3c936125bfb : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/002dcf04510e4034bb29dbe88a822721 : SUCCESS in 36m 22s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/40b0c96f98e948d4a17fe453bfb4f68e : SUCCESS in 1h 16m 13s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5c9490cccfc948269b48e9fb4852ea9d : SUCCESS in 1h 09m 10s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2ebe0728e17648ffb01b0ae73cf70a93 : FAILURE in 16m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d2555e129f64f6f9b5b55c4eeb656c5 : FAILURE in 11m 15s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7cdb3368d69f448b9d0a6bbf34700fb5 : FAILURE in 1h 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c82e4bd26e05e908c0415f8bf6f4f71d096c6809","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-22 23:19:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c594d7c7e59c690499cacc96c4227f52460006f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 00:47:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0d0175e5b81447094c6514b6efd94f7 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e3a55a385cf44f6a7f19effba650fe5 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f78a847ad5b4149be5d4671b1ceb465 : SUCCESS in 4m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f95fbd5b4f6049deba3a60f140dd8257 : SUCCESS in 16m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f5874c8f44e48c1b22f9c3eff584102 : SUCCESS in 10m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/205d7cd6c2d0494fa93c23b448dd2984 : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2af91c694f484a358f20283a984ebdb1 : SUCCESS in 41m 56s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/470baaca562b4a1bbd4dc63ac9eb632d : SUCCESS in 1h 03m 29s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aeade8108eb94b29897893133da33ee5 : SUCCESS in 59m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1af3fac4c49746e3aaaff94229f9b553 : FAILURE in 15m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f88ffa137e494000b22f033dce491aaf : FAILURE in 13m 29s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f842c7dafe87442dac5c46caf6e91104 : FAILURE in 1h 22m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"144b3bf4bc9b9df8c7dbf3f8d421e7ee14d7b251","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-23 03:37:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8263eba30f9ec72b99e3bf7c84b3a3a7d1672c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 05:04:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08bd0e605b0c458685d12abaf1737809 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50736a0d1fc34c38bfe4eecfbf9b258f : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/604580c832f146ae83b63cc0abd4171e : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/012a06ed80ea4691bf8ad221631526c7 : SUCCESS in 14m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bad8441ef0d4ef99f7a24ab8983b3ff : SUCCESS in 9m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49e3d9f7a4eb4fd3b3124107f0808a48 : SUCCESS in 10m 49s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51ef32c489f7481dac147725f383999e : SUCCESS in 36m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5f35a86d4dc3480487b32208de963001 : SUCCESS in 1h 01m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0b0dcef93a674b279f0f487d262f83ea : SUCCESS in 1h 01m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d2838966f512459398c58e15a847b312 : FAILURE in 13m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0636aa3a88414fd0aa7bb3682d987f30 : FAILURE in 11m 33s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/354a7cd9eda24155bbd7864076ffdaa4 : FAILURE in 1h 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e2e248ed205c7c70bb47860efc2f43c7d33c2ec","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-23 13:57:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"740d7cf6290c15a7b9c21e1964330802d0d095de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:22:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ad2fb6078d545eb9aea9c51798eb29f : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04435a991e8b4367ab7f62ebe9738cc9 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1ea0c54dbdf484088a1d14ecaf30992 : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/108c3bfbf439456090cd96fb7db38b72 : SUCCESS in 18m 56s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4757eaf9eb3422ca01d3a6104a351f1 : SUCCESS in 12m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db33c3f1f2614f158af30a253573069c : SUCCESS in 10m 51s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/767fd23e817c48fc990070788288e645 : SUCCESS in 42m 42s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/18b6f5591782420f9f2124460cc44324 : SUCCESS in 1h 13m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/597286490a7a4e0cb5f4f8bed597a42c : SUCCESS in 1h 00m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/29fe2287891f4377a3ff71c12cc42289 : FAILURE in 14m 56s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04e82657d06d4716bc8ae7a0a5fb530c : FAILURE in 16m 17s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8dbe9a29c32e4686be47219c298210e8 : FAILURE in 1h 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be4a49619dda4513e6b9319d99d8065cda991187","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-24 02:59:51.000000000","message":"Patch Set 1:\n\nThu Jan 23 14:15:50.632645 2020] [:error] [pid 2599] [remote ::1:0]   File \"/usr/lib/python2.7/site-packages/heat_dashboard/api/__init__.py\", line 34, in \u003cmodule\u003e\n[Thu Jan 23 14:15:50.632679 2020] [:error] [pid 2599] [remote ::1:0]     from heat_dashboard.api import heat\n[Thu Jan 23 14:15:50.632698 2020] [:error] [pid 2599] [remote ::1:0]   File \"/usr/lib/python2.7/site-packages/heat_dashboard/api/heat.py\", line 15, in \u003cmodule\u003e\n[Thu Jan 23 14:15:50.632732 2020] [:error] [pid 2599] [remote ::1:0]     from urllib import request\n[Thu Jan 23 14:15:50.632770 2020] [:error] [pid 2599] [remote ::1:0] ImportError: cannot import name request\n\nurllib does not have a request function in py27\nhttps://docs.python.org/2.7/library/urllib.html\n\nPython 2.7.17 (default, Oct 20 2019, 00:00:00) \n[GCC 9.2.1 20190827 (Red Hat 9.2.1-1)] on linux2\nType \"help\", \"copyright\", \"credits\" or \"license\" for more information.\n\u003e\u003e\u003e from urllib import request\nTraceback (most recent call last):\n  File \"\u003cstdin\u003e\", line 1, in \u003cmodule\u003e\nImportError: cannot import name request","accounts_in_message":[],"_revision_number":1},{"id":"13e19bdad46f93ccd2192a7b7d6f394e5e75d9dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-01-24 09:21:43.000000000","message":"Patch Set 1:\n\nRDO will pin commit on that pkg and that should fix it until we are swapped over completely to centos8","accounts_in_message":[],"_revision_number":1},{"id":"7147bc658218133efbf407a1f02909c3b489c58a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-01-24 09:27:50.000000000","message":"Patch Set 1:\n\nhttps://review.rdoproject.org/r/24659","accounts_in_message":[],"_revision_number":1},{"id":"04cfd4ea4e1512e12d6fc100fae7775ff4fc0b38","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-24 13:27:09.000000000","message":"Patch Set 1:\n\nreview","accounts_in_message":[],"_revision_number":1},{"id":"8c592070fe0711d48f58719879ed9199766eea60","author":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"date":"2020-01-24 13:33:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2da8f2c45ac78f14ede6aff3c4201d39a5adb9cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 15:20:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1f58316dcd74d12b1ef03916906f000 : SUCCESS in 6m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6beab92d93547adb345e0f14a18adec : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57066adce989488aab77fa10656af8cf : SUCCESS in 6m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/312c2779295a44549bb378e1e22b7193 : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c62a607ace13463d81448f550ad051e7 : SUCCESS in 13m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4fdf7d8050c4e7badeb58610ae211a6 : SUCCESS in 11m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/853e88f941694f428ff72ab3535b2abb : SUCCESS in 39m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da6ef6e611b44daba5f8584c0f13f2ec : SUCCESS in 1h 01m 05s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/098b1f469c4f4d57a7a734a0fbbb5a17 : SUCCESS in 1h 02m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6fe2c6281171412ab4de53edf6c0fef5 : FAILURE in 17m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bfd7efc3fb849b7aa7b837fad059204 : FAILURE in 19m 11s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6135c797b6c5466db8767d8b3a2559a9 : FAILURE in 1h 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7597674f9b737c8aa0d6705fd52345aa5a119aa8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-01-24 19:40:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f766a4dc8b22cf561fb98bff1bca7034952d70a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 21:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/835f41db46ec41738f5b4ad8a88e0aff : SUCCESS in 8m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/107d630682634d6b94df4adf46574d76 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04c9aff2150547a89af42bba36fb6ae6 : SUCCESS in 5m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/03f026797d9941ee8b16671612abfacd : SUCCESS in 16m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/556ea756f16746289dee93547ca2fa9a : SUCCESS in 15m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af6dd120c6534d0f817d59ac1da21337 : SUCCESS in 16m 19s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e39b9802967495fa040e70e2b6d27cc : SUCCESS in 49m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9e70ad309d8d46e1955f6d484c7cf8c7 : SUCCESS in 1h 13m 48s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/962ac2c331b241d0bdf3a042cd75ad55 : SUCCESS in 1h 22m 19s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/499e2b7168074f10b6b82f72630e8bae : SUCCESS in 14m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4aa67b118ad44395a845b8ba3bd12e10 : FAILURE in 12m 48s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f97f6adeab9e471495bb9e4791d3eb4d : FAILURE in 1h 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d92fed744d89e98a6133815602fd8e366bfb0fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-24 21:57:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd3bb34c91d2a12fde25cb4c20704facc73af9d7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-01-25 13:41:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3bab22e72cc175c840190ada8f1bf71bbda35a59","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-27 14:48:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93c8e10c1a3d3601870d25f655c134b2c06bd08e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:11:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56b2021f2afa4a98b7074761225bd9ec : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40856ffef4ec45c4a16b3072d8053916 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e867966f1004081af8a71c67977cb5f : SUCCESS in 4m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d876b8f50c1542d38bbfa2edeb962377 : SUCCESS in 15m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/999c58b73d444b0fb8d5175942a2f962 : SUCCESS in 12m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/056030c654a841e398282c7361970e55 : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57c2760d6d6a49c3b9beb37fd6057361 : SUCCESS in 50m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4100c0cfd16f4699b4a2a38c6364b124 : SUCCESS in 1h 03m 57s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fe74ef6ae5834e838976498a397c20c5 : SUCCESS in 1h 01m 53s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/eb992954385445a9a2e2d274f78ca4f2 : SUCCESS in 18m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e7706782a474973b3f2c6250e0659b3 : FAILURE in 16m 54s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b999da777e4f4a058875cf08dbf6dcfe : SUCCESS in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"681d91b4bfaac9725e8a6ecb2ac4b224049d3f75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-01-27 16:23:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1677c361b94d52284efc364e53a06b6d7e33b8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 16:23:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dffa0e15cb18cdd4f5465a2902d573246cb624c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:03:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/605abf6ece5541e59da66e4362825aa2 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8acdc71b5b394f2e80a2fab759cf08bc : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3244f3b335c24071b9c9caaf7dee459e : SUCCESS in 6m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2bffd89301c64046acefdf05223bc3a6 : SUCCESS in 19m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82a1394b6e8540c9a3d1fc4e10fd6cfe : SUCCESS in 12m 22s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4922a18b6d11407eb7b7e319c05958c7 : SUCCESS in 1h 04m 05s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/063ee6f1d50e4879807d7170b3fa3810 : SUCCESS in 24m 12s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ba82ebc29d9b45a8a949c4c75e42d5a5 : SUCCESS in 1h 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6f937a51f40741df5dad1bccd91b50b7279ad73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:03:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc6226565b9f4d4865e6ced0e001514521315fd9","revisions":{"fc6226565b9f4d4865e6ced0e001514521315fd9":{"kind":"REWORK","_number":1,"created":"2020-01-22 21:27:16.000000000","uploader":{"_account_id":25820,"name":"Marc","email":"mb.methot@gmail.com","username":"mmethot"},"ref":"refs/changes/71/703871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/71/703871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/71/703871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/71/703871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/71/703871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/71/703871/1"}}},"commit":{"parents":[{"commit":"d0639c29ec11b1d16d114f38bd5388941150889c","subject":"Fix stdlib version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d0639c29ec11b1d16d114f38bd5388941150889c"}]}],"author":{"name":"Marc Methot","email":"mb.methot@gmail.com","date":"2020-01-22 21:25:53.000000000","tz":-300},"committer":{"name":"Marc Methot","email":"mb.methot@gmail.com","date":"2020-01-22 21:27:15.000000000","tz":-300},"subject":"Adding the HTTPONLY cookie header to secure_cookies","message":"Adding the HTTPONLY cookie header to secure_cookies\n\nChange-Id: Ic34170530b260b426fd65ac96aa5f494591c2ff1\nCloses-Bug: #1860608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/fc6226565b9f4d4865e6ced0e001514521315fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/fc6226565b9f4d4865e6ced0e001514521315fd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
