)]}'
{"id":"openstack%2Fpuppet-horizon~710484","triplet_id":"openstack%2Fpuppet-horizon~stable%2Ftrain~Ife921f8409f58b70c5da4c344e330eec56669e20","project":"openstack/puppet-horizon","branch":"stable/train","topic":"no-autoindex-stable/train","hashtags":[],"change_id":"Ife921f8409f58b70c5da4c344e330eec56669e20","subject":"Reflect openstacklib options for apache::vhost resources","status":"MERGED","created":"2020-02-28 16:37:50.000000000","updated":"2020-02-29 00:26:13.000000000","submitted":"2020-02-29 00:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710484-1582935973273-b95df229","meta_rev_id":"0cedbb3cf6dd707f7c1a0c30a2f350d11ed5aaf3","_number":710484,"virtual_id_number":710484,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-29 00:26:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-28 22:56:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-28 22:56:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-28 16:37:50.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 16:37:50.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-02-28 22:56:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-29 00:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"799872c4bcfaff3088be38b84a061d5d66b3cb70","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-28 16:37:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"149dab0e5f3873b1ba83b13736572e36da733531","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:20:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a537d26f32c4ad9883d41c677c2ff57 : SUCCESS in 6m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06342635ff1640c983d5c10a034b7f6f : SUCCESS in 7m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0b65bb965804c9b9cb008a7e20e2a75 : SUCCESS in 7m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0b35397316b44be1ae6a8aca9205b4f0 : SUCCESS in 20m 06s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/933bb241379c4dedb658ff6692a526a6 : SUCCESS in 38m 04s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f5df50456cfb450dbfee773b91526deb : SUCCESS in 1h 00m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6a237668c93647a291cc9cbfbef8d3cb : SUCCESS in 17m 58s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/082209f933404c48840e9879d18802c1 : FAILURE in 15m 50s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5fe88cd1165d4679a8a27bc2179aac3d : FAILURE in 1h 43m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2d02027144da02fe6cc0be28f8b02615a22194f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-28 22:56:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"903c1282ada5f24f01c2a4e7dc0972342fa3388b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 22:56:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"018e6c6efa2fbdcf2403ce5b6bb6a3acde5787f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 00:26:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2943156bec4d0ef7b0789b36c6ad58dfdec26800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 00:26:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb47214fdd684906ab0b410d5a9f0313 : SUCCESS in 6m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2446bc584e984a77851046a5364f43fe : SUCCESS in 5m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71caa404428747fda1e97cab58265baf : SUCCESS in 6m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/071e3fa47f674fdfac1c74e6f3f589ad : SUCCESS in 17m 38s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/38c7c0290e4d4e39a2c1afda1bca45e9 : SUCCESS in 1h 07m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ffa7cde4cecf489fb74418c3f71d44bd : SUCCESS in 16m 28s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21cd11f6e5c54e5389d731df3cdbc39c : FAILURE in 1h 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cadfc23c55bc1fbc81735ac292b58cdb0b953d3","revisions":{"3cadfc23c55bc1fbc81735ac292b58cdb0b953d3":{"kind":"REWORK","_number":1,"created":"2020-02-28 16:37:50.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/84/710484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/84/710484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/84/710484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/84/710484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/84/710484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/84/710484/1"}}},"commit":{"parents":[{"commit":"620e8f62fe02cdc8fbb13f7aaf68c9609fda3155","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/620e8f62fe02cdc8fbb13f7aaf68c9609fda3155"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-28 07:08:26.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-28 16:37:50.000000000","tz":0},"subject":"Reflect openstacklib options for apache::vhost resources","message":"Reflect openstacklib options for apache::vhost resources\n\nIt was discovered that Directory Index was enabled by default, leading\nto potential leak of data.\n\nA patch was done in puppet-openstacklib[1], but horizon doesn\u0027t use this\ncommon lib, meaning it wasn\u0027t properly fixed.\n\n[1] change id Icba53f4e32237556608f4cb6dcd9da1a71705c19\n\nRelated-Bug: #1854442\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1729423\n\nChange-Id: Ife921f8409f58b70c5da4c344e330eec56669e20\n(cherry picked from commit 7ec0c090ebdce9f48dbfa073c9427761c1a5ef11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/3cadfc23c55bc1fbc81735ac292b58cdb0b953d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/3cadfc23c55bc1fbc81735ac292b58cdb0b953d3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
