)]}'
{"id":"openstack%2Fpuppet-horizon~710486","triplet_id":"openstack%2Fpuppet-horizon~stable%2Frocky~Ife921f8409f58b70c5da4c344e330eec56669e20","project":"openstack/puppet-horizon","branch":"stable/rocky","topic":"no-autoindex-stable/rocky","hashtags":[],"change_id":"Ife921f8409f58b70c5da4c344e330eec56669e20","subject":"Reflect openstacklib options for apache::vhost resources","status":"MERGED","created":"2020-02-28 16:38:14.000000000","updated":"2020-02-29 00:26:13.000000000","submitted":"2020-02-29 00:26:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710486-1582935973880-28c1a561","meta_rev_id":"ff10738c5187d297edcc7557b243bffd8f077240","_number":710486,"virtual_id_number":710486,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-29 00:26:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-28 22:56:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-28 22:56:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-28 16:38:14.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 16:38:14.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2020-02-28 22:56:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-29 00:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54d0f0c1b7af21b884b350999c9ac758c2f85cc2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-28 16:38:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4f9f7e035b19c8ee7a475312ce09db65ff6236f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 19:09:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/8efb109582da4b1cb52865375c18abaf : SUCCESS in 8m 14s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/8ca2b4ed2f0a48abb8c266fb9f70b2af : SUCCESS in 8m 25s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/6c2e6721105e40118946a5ce467e124f : SUCCESS in 7m 44s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/d61c69adfc5e4787ab611c48a14235ac : SUCCESS in 18m 20s\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/630e051bfc4746549b4e6c321b4a6526 : TIMED_OUT in 1h 31m 35s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1f076af70a1d4b01b17e10f80754d4dc : SUCCESS in 1h 13m 32s\n- puppet-openstack-integration-4-scenario003-tempest-debian-stable https://zuul.opendev.org/t/openstack/build/c659e8260607472b85c2497626891d77 : TIMED_OUT in 1h 31m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bc92d4c0a3824674bff43256452f7e90 : SUCCESS in 21m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94a194a580284af5afe5393cdbab72cc : SUCCESS in 23m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a5e5d0e34c181ee95edc60b3c63761f1876248fd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-28 22:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cdac8a11a39175eff098a6e09d81b869149239bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 22:56:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe517ec4c25fd1ce7c594d2a15b938f313c8e32b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 00:26:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e98907f057d348108fccddc273bf3a33c0833d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-29 00:26:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/7cf719aee9e9490b93a4a82d8973127a : SUCCESS in 7m 25s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/b9548e0aa8164248b33cf0d93f9ee675 : SUCCESS in 6m 47s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/4b682834fe6a4b059b1965db1e451b9a : SUCCESS in 6m 39s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/65dbb27fe20946d4816dda681422048c : SUCCESS in 18m 11s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/13eb5cab06664acf8d0589acdf8c3707 : SUCCESS in 1h 07m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e7d422f600614f2b984ccd2653fb1d73 : SUCCESS in 22m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56abcaf42bbe4886b0b37eaf2355f201 : SUCCESS in 21m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bacf1cb9e5b8c914b1dc5826b4b9b4f749a13160","revisions":{"bacf1cb9e5b8c914b1dc5826b4b9b4f749a13160":{"kind":"REWORK","_number":1,"created":"2020-02-28 16:38:14.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/86/710486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/710486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/710486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/710486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/710486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/710486/1"}}},"commit":{"parents":[{"commit":"a0f8f780267f10f929d17ec891b83740581f6cf3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/a0f8f780267f10f929d17ec891b83740581f6cf3"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-28 07:08:26.000000000","tz":60},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-02-28 16:38:14.000000000","tz":0},"subject":"Reflect openstacklib options for apache::vhost resources","message":"Reflect openstacklib options for apache::vhost resources\n\nIt was discovered that Directory Index was enabled by default, leading\nto potential leak of data.\n\nA patch was done in puppet-openstacklib[1], but horizon doesn\u0027t use this\ncommon lib, meaning it wasn\u0027t properly fixed.\n\n[1] change id Icba53f4e32237556608f4cb6dcd9da1a71705c19\n\nRelated-Bug: #1854442\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1729423\n\nChange-Id: Ife921f8409f58b70c5da4c344e330eec56669e20\n(cherry picked from commit 7ec0c090ebdce9f48dbfa073c9427761c1a5ef11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bacf1cb9e5b8c914b1dc5826b4b9b4f749a13160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bacf1cb9e5b8c914b1dc5826b4b9b4f749a13160"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
