)]}'
{"id":"openstack%2Fpuppet-horizon~711900","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fqueens~Ife921f8409f58b70c5da4c344e330eec56669e20","project":"openstack/puppet-horizon","branch":"stable/queens","hashtags":[],"change_id":"Ife921f8409f58b70c5da4c344e330eec56669e20","subject":"Reflect openstacklib options for apache::vhost resources","status":"MERGED","created":"2020-03-09 10:53:05.000000000","updated":"2020-03-09 16:06:34.000000000","submitted":"2020-03-09 16:06:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711900-1583769994940-77b288b9","meta_rev_id":"6541942bdfc97d5659da294b81f87be72c32194b","_number":711900,"virtual_id_number":711900,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 16:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 14:54:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-09 13:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-09 14:54:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-09 11:59:25.000000000","_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 11:59:25.000000000","updated_by":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"reviewer":{"_account_id":25613,"name":"Keigo Noha","email":"knoha@redhat.com","username":"knoha"},"state":"REVIEWER"},{"updated":"2020-03-09 13:59:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-09 14:54:20.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-09 16:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a52596c7b742b85c3aa3cbea120650e9c47b3ce3","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-09 10:53:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"930af6919ec8d70a7ae297d32c09d706667ccd75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 12:25:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6a0f9962219d40c08cdc89b3a42d89ab : SUCCESS in 4m 47s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/4b8ab2bc569b475894f8287039e72095 : SUCCESS in 5m 22s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/d9afe2074a174c549a981348c7945dbd : SUCCESS in 14m 43s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/590607bce26c42a2a7a0626c83e741c4 : SUCCESS in 15m 52s\n- puppet-openstack-integration-4-scenario003-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d028b5257af942518beadf07b9341611 : TIMED_OUT in 1h 30m 59s (non-voting)\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e1d4e1e0a3584727b69593cb6a0f86d0 : SUCCESS in 1h 00m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cc5f2fc479b149b8a0e0f01efb7e74cb : SUCCESS in 15m 53s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/9f0e4582b251425685d38ad234e26e64 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ea98bad187e5e8c2dc5dcc97841093ad9954481a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-09 13:59:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df716b8e2caa6517394e1856faa1eea481b85011","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 14:54:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"451ec4e2702971aaf85fce880d3225fc355034ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 14:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1719c95d6563f23f1b56866795ce429feecfe81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:06:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/57b8d8653d504de69bd67746b65b357b : SUCCESS in 5m 11s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/830119fc176f416283b267f6e2a9f684 : SUCCESS in 4m 38s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/d3be7b52b9bf424181229aa122d3b025 : SUCCESS in 23m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/206e9f19ced746f383965c43aeef8192 : SUCCESS in 13m 18s\n- puppet-openstack-integration-4-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/63e95e5c7c1847dbacc9c12be90ed001 : SUCCESS in 1h 09m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee0a4f3c59594c9ea7176e1aeca6e6f0 : SUCCESS in 15m 53s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/74ef25428e4c418d8a5b3cce2565a5f8 : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6541942bdfc97d5659da294b81f87be72c32194b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:06:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a91c7d22d3ba88dd7a488e8d26145ccc294571c","revisions":{"5a91c7d22d3ba88dd7a488e8d26145ccc294571c":{"kind":"REWORK","_number":1,"created":"2020-03-09 10:53:05.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/00/711900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/00/711900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/00/711900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/00/711900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/00/711900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/00/711900/1"}}},"commit":{"parents":[{"commit":"c6c5d6692ee97ecf9bcf5250d193791ad8d518d2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c6c5d6692ee97ecf9bcf5250d193791ad8d518d2"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-28 07:08:26.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-03-09 10:52:31.000000000","tz":60},"subject":"Reflect openstacklib options for apache::vhost resources","message":"Reflect openstacklib options for apache::vhost resources\n\nIt was discovered that Directory Index was enabled by default, leading\nto potential leak of data.\n\nA patch was done in puppet-openstacklib[1], but horizon doesn\u0027t use this\ncommon lib, meaning it wasn\u0027t properly fixed.\n\n[1] change id Icba53f4e32237556608f4cb6dcd9da1a71705c19\n\nRelated-Bug: #1854442\nRelated: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1729423\n\nChange-Id: Ife921f8409f58b70c5da4c344e330eec56669e20\n(cherry picked from commit 7ec0c090ebdce9f48dbfa073c9427761c1a5ef11)\n(cherry picked from commit bacf1cb9e5b8c914b1dc5826b4b9b4f749a13160)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5a91c7d22d3ba88dd7a488e8d26145ccc294571c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5a91c7d22d3ba88dd7a488e8d26145ccc294571c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
