)]}'
{"id":"openstack%2Fpuppet-horizon~730486","triplet_id":"openstack%2Fpuppet-horizon~master~I9e12762538e70bcf838f4626c8ea9e3489bb7014","project":"openstack/puppet-horizon","branch":"master","topic":"multiple_log_handlers","hashtags":[],"change_id":"I9e12762538e70bcf838f4626c8ea9e3489bb7014","subject":"Allow multiple log handlers","status":"MERGED","created":"2020-05-24 01:14:15.000000000","updated":"2020-05-27 20:09:03.000000000","submitted":"2020-05-27 20:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730486-1590610043965-ad19e745","meta_rev_id":"256162d675d1d01e8f70ffc42f295dee8bf7a4f6","_number":730486,"virtual_id_number":730486,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-05-26 12:04:33.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 20:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-26 21:09:16.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-27 17:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-27 17:26:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 02:28:28.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-05-26 21:09:16.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-27 17:26:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-27 20:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a11b2993723aaa977db5015afee03607f02f92f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23d0b9d71391182aea8df8ad5f328ea43d6c7bac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:15:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69bbe440fb706065e8bef5225dcdb2bf9d1dffc6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:15:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb3bddda652f944e74935593be0d50eb127c9ac4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-24 01:56:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"595573ff4ddbb4da89d66bfae985d38a3673d26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 03:24:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c8e93cdac434e0cafc6cddf8430fc7c : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4970624d0018415fb801dbc9e3a05035 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f61ecd2c82444e5c96c42412fb99f0e9 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a718a0db01c54e2d967041a45634bfab : SUCCESS in 17m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6482edec04b743e0b575461642b0b2e8 : SUCCESS in 18m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cb524f10da147daa6f084341938c5e7 : SUCCESS in 16m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b55a3995a8ce4749aa7e59502f3af8d7 : SUCCESS in 11m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc328e868a534d3197ce95f16ac1474c : FAILURE in 47m 56s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2bd0a253d1f64b23940dd096177b9a77 : SUCCESS in 55m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f69b454c281c4dcc8346458ccd7c3647 : SUCCESS in 14m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d3acfdedd65d4d8597b4dda19e9d08dd : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1df91154d4b84c7f9fea93394a74608c : FAILURE in 7m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df7e8aa04de74268b85a30c035c14af7 : SUCCESS in 3m 18s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ea4cfa54a4b24c569a278feaee74ad73 : SUCCESS in 1h 22m 59s","accounts_in_message":[],"_revision_number":4},{"id":"b1036cfe581f213acecbb0d7fddfa90674043292","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-25 02:24:53.000000000","message":"Topic changed from mltiple_log_handlers to multiple_log_handlers","accounts_in_message":[],"_revision_number":4},{"id":"f71ea0dd2ea5d87e6e4a47789e042348e2bbf6f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-25 02:25:44.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"36e312fe8dfb2ed76d3862e8ae96d06d541a7747","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-05-25 02:28:28.000000000","message":"Patch Set 5: Code-Review+1\n\nI like this idea :)","accounts_in_message":[],"_revision_number":5},{"id":"ebf81810921614cedbf94b32013377dd57781da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 03:54:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0339a59b975148d9b6bb05a5fbbcbbc8 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb0decbba8664b9f8c40af79740d4cf6 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45226cdc61624a9780c633fdacde9b5f : SUCCESS in 4m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a201faecbb4e4599b8d15893b5dc510b : SUCCESS in 12m 30s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ab3b387a6d704fba852f1f60f736d062 : SUCCESS in 13m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8282385e2a27409eb798adef1ffd4e16 : SUCCESS in 11m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c78daab0a0454e758ffbc6a45283506a : SUCCESS in 11m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22a0fd780f494496a78433c38c283fe2 : FAILURE in 33m 22s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f21aad60aa5946b79c07851c3b06dbf8 : SUCCESS in 1h 17m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/af15c0c171f841519cced322918a8052 : SUCCESS in 13m 59s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e9b8ff61240444eeb8fe244eacb2049e : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0b30418119d4defaf76785145e72b05 : FAILURE in 5m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6485c8e12f734c0297ee79676d36d0c4 : SUCCESS in 3m 19s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7b5e214799b149958475b1bb0dabd4a6 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":5},{"id":"50f6b81cdac33cd326493fb9b0e29f5ba2ec6ab1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-25 06:56:59.000000000","message":"Patch Set 5:\n\nDoes it really support settings log_handlers in local_settings.py to [\u0027console,file\u0027] etc?","accounts_in_message":[],"_revision_number":5},{"id":"aec6afc7dfbe839e4a7bdcededb97890a5ca0d1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-25 07:02:51.000000000","message":"Patch Set 5: Workflow-1\n\nThanks, Tobias for your good catch.\n\n\u003e Does it really support settings log_handlers in local_settings.py to [\u0027console,file\u0027] etc?\nIt actually supports multiple handlers, but it should be described like;\n [\u0027console\u0027,\u0027file\u0027]\n\nI\u0027ll update the patch to have correct format. might need some change in unit tests to cover this...","accounts_in_message":[],"_revision_number":5},{"id":"41139f979de6b7a207764d38c35f59caa4a32b50","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-25 07:10:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"288154e86bee5a5147d5586086a7c2d0c6e46ff6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-25 07:13:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8032f36e520730fc60a386a58fd3f12e3368bd86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 08:38:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d35d61a187ad41868e4d65b605363b95 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47bfe1c7192c4cf881457ac9835f56c6 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/312dbf3612d9406f93178fbbfcafe9dc : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b25f338c87e242f9b47383905bf83675 : SUCCESS in 11m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cebebeecd7cd4b04899862e28ba818da : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efc76477b0364a5982cbc26a4568e391 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55b48a077ca243809ba6b53cd0afb684 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5420008ead4c4c3a8a3af191c08d9f7d : FAILURE in 33m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/02834306d29a438da90a74925880f978 : FAILURE in 36m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b9fe5540bead46e988a81737baf9d3a9 : SUCCESS in 15m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/255ddc3a770e46b587fee3dc098726e5 : FAILURE in 10m 07s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28dcaeffdb5c4c04a26d31269b654aae : FAILURE in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d1cd3dab9b44a31a7a2602ff744ed46 : SUCCESS in 3m 32s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/00fbb0b46f474d469a5270bffdb6ac4f : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":7},{"id":"e64229c5e3c77c9bd703b4188b27591960b50410","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-26 12:04:33.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"7eaa0e29e91b7968c9cae83a409ec716c4cd5134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:31:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6fc5fe97503423eb7c108ab1b8c6b15 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c75e31439a54da78ed03e324fd243c0 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e25dd69a8214d28bd03e71ac6f18295 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6b2b71c42cfe4e549d1cb3e829e8b778 : SUCCESS in 15m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f4eb7105c9d840668b99449245139f73 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5e52425b9a94e6e918baa09cb01c7c9 : SUCCESS in 12m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123ad72e357b46bd8cabf7dda20785d0 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7410dc471a8b46698c35a48c1377a00f : FAILURE in 34m 29s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/26530b459d7e46248835b7197e197850 : SUCCESS in 1h 07m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b97fc8adcca6407db2c42058756a9470 : SUCCESS in 17m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3b38cc7983d7480c8ef599b235c8c971 : FAILURE in 11m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c773b53a19f496f9882de64ee0425df : FAILURE in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b4c33ce88a544188e6ca5b85cf4c159 : SUCCESS in 3m 45s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/81c94f38cbd94ac594ee8a7eee95ea4b : SUCCESS in 1h 24m 38s","accounts_in_message":[],"_revision_number":7},{"id":"c4ade740d6753c4ec841362589c120d5b90e7306","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-26 21:09:16.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0c1c3fd92b210df1db00642afcdf5005bdccd586","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 17:26:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e7153e7142cfc572588ac6844ad4f75d69a5378c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 17:26:28.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e7f6e32349c0cc9b154dd1d6ae0426fa73699504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:07:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d1ba21bd265a370b43faa1c491fed747f7aa1a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:07:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e8fa8292c8e461a99968706d88c5496 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08a6f44dbffc48bfadca40453a8c0d62 : SUCCESS in 5m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1131dc599f51402bb168143ad4946377 : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d120e8171f694291b71b7e1aab7f64ea : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fc64ee491e8b4a05ae250003d63ce7ab : SUCCESS in 15m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/736bf15e673241649e7f2114522dab7d : SUCCESS in 10m 59s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a01c50b038fa47f59fcacd508bb864d8 : SUCCESS in 1h 19m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f7b3a898b4004767a335b9dd6d600146 : SUCCESS in 18m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5a35c9e6754ffcb3c772b8d41ffa5f : SUCCESS in 4m 11s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b3a5ef40da24ba3bcdb20597ee3d9e8 : SUCCESS in 1h 35m 15s","accounts_in_message":[],"_revision_number":7},{"id":"256162d675d1d01e8f70ffc42f295dee8bf7a4f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 20:09:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad7c349701fb4099888a25aba0082b42 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dfb3cc23c67a4435c4802ee01dbc4932f9f70e3a","revisions":{"6816d117511b464267a3396481dd43bc3bc26219":{"kind":"REWORK","_number":1,"created":"2020-05-24 01:14:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/1"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:14:05.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6816d117511b464267a3396481dd43bc3bc26219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6816d117511b464267a3396481dd43bc3bc26219"}]},"branch":"refs/heads/master"},"8f288342e4dcd60ee11660e65d04c32d4b6abacb":{"kind":"REWORK","_number":2,"created":"2020-05-24 01:15:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/2"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:14:59.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8f288342e4dcd60ee11660e65d04c32d4b6abacb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8f288342e4dcd60ee11660e65d04c32d4b6abacb"}]},"branch":"refs/heads/master"},"032bf11812b281e1af59dba9c39cbb5724aaba5c":{"kind":"REWORK","_number":3,"created":"2020-05-24 01:15:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/3"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:15:28.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/032bf11812b281e1af59dba9c39cbb5724aaba5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/032bf11812b281e1af59dba9c39cbb5724aaba5c"}]},"branch":"refs/heads/master"},"ec8eeb8f25ae0e7a8dea3d730a74c4a571553219":{"kind":"REWORK","_number":4,"created":"2020-05-24 01:56:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/4"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:56:47.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ec8eeb8f25ae0e7a8dea3d730a74c4a571553219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ec8eeb8f25ae0e7a8dea3d730a74c4a571553219"}]},"branch":"refs/heads/master"},"14d0f5c809afdf379d4134acf1b5196a499d3784":{"kind":"REWORK","_number":5,"created":"2020-05-25 02:25:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/5"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-25 02:25:41.000000000","tz":0},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/14d0f5c809afdf379d4134acf1b5196a499d3784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/14d0f5c809afdf379d4134acf1b5196a499d3784"}]},"branch":"refs/heads/master"},"47b5530ce7d823df3d536135cc4e45560c480182":{"kind":"REWORK","_number":6,"created":"2020-05-25 07:10:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/6"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-25 07:10:17.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/47b5530ce7d823df3d536135cc4e45560c480182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/47b5530ce7d823df3d536135cc4e45560c480182"}]},"branch":"refs/heads/master"},"dfb3cc23c67a4435c4802ee01dbc4932f9f70e3a":{"kind":"REWORK","_number":7,"created":"2020-05-25 07:13:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/730486/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/86/730486/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/86/730486/7"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-24 01:10:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-25 07:12:53.000000000","tz":540},"subject":"Allow multiple log handlers","message":"Allow multiple log handlers\n\nThis patch allows operators to add multiple log handlers for horizon,\nso that operators can define multiple outputs.\n\nChange-Id: I9e12762538e70bcf838f4626c8ea9e3489bb7014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dfb3cc23c67a4435c4802ee01dbc4932f9f70e3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dfb3cc23c67a4435c4802ee01dbc4932f9f70e3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
