)]}'
{"id":"openstack%2Fpuppet-horizon~731356","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fussuri~I0a76cbe45774bbf6871c1803c056bc2f6b3b6c0d","project":"openstack/puppet-horizon","branch":"stable/ussuri","hashtags":[],"change_id":"I0a76cbe45774bbf6871c1803c056bc2f6b3b6c0d","subject":"Consolidate python memcache bindings install","status":"MERGED","created":"2020-05-27 21:48:46.000000000","updated":"2020-05-28 14:23:38.000000000","submitted":"2020-05-28 14:23:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731356-1590675818003-60d2867d","meta_rev_id":"77bd34caaee46ef68493af29f2449ceef5401691","_number":731356,"virtual_id_number":731356,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 14:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-28 12:56:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-28 07:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-28 12:56:15.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 07:46:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-28 12:56:15.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-28 14:23:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c89d6e4bf8d08d7457037098cdee87532db9f6f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 21:48:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c38d39b88b8164f288428cb5ed1a8b21e3e8b7b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 21:48:49.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 74a4f7e023d2e8e62d48a762487dadb007a73498","accounts_in_message":[],"_revision_number":1},{"id":"c59948cf9f91114a97c9c6e57180fb977da8cd82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 23:14:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c402ac01b0e54cf9900559cfd9b3d3b4 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86a275d990dd4ba181e9fbbca1bbf8b2 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31a96c9125de4053baa941a657af7c19 : SUCCESS in 6m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a9112df9131648fc9decd5e79e431c99 : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0858b796fd6643bf91e1c789b7b08d7d : SUCCESS in 12m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96d4d24eff83449ebe8d82d79f905723 : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f5ef4d6ac794863b546a0b6bd5ae442 : SUCCESS in 12m 32s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fda6a0bab46b44a981d47581928976fc : FAILURE in 32m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/80b414edbb434cd7a67374ff60cf6b9f : SUCCESS in 1h 08m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7a202c1ae2ef4a549e7252ca6b5f12a6 : SUCCESS in 18m 18s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a4152d0fcfe644559119865e0435df4d : FAILURE in 11m 14s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d334ea8a92d740338bec8e67abdee210 : FAILURE in 7m 00s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3988eb32098c4aaaa568489fd3f9a1d3 : SUCCESS in 1h 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"085979c0dc2ee4834dbea8c4e5fa91f114e15964","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-28 07:46:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c64ea7648ecdf40e83386a39061e78958fc392d9","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-28 12:56:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45716eb41315a049b61f6ac9f1e884961211461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:56:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebe515f00fe3f46b813f6bc74f2f87ebbc6d0a65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:23:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc0d9259c2354afda9080dae1312af0e : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b0040f3d1a04fb09f2e21cb3296ebd5 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df3b0cdccf3c4518afe526319c182d9d : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/32ed1df442e744efa2b9bf098a5aabe1 : SUCCESS in 14m 59s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b66ec5408df240ad8ecc8740d5e42144 : SUCCESS in 14m 24s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a9312883b764287a820888f774bf518 : SUCCESS in 10m 54s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c955a5c8c8434853be63e50432881a26 : SUCCESS in 1h 17m 57s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/13f8985e31ae4898b4b1731d9b16b94f : SUCCESS in 15m 48s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab6fa66c76a141189b55554afb7faf16 : SUCCESS in 1h 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77bd34caaee46ef68493af29f2449ceef5401691","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:23:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c8e18b5baba4c427eda4d8f505907a89ccd195d","revisions":{"9c8e18b5baba4c427eda4d8f505907a89ccd195d":{"kind":"REWORK","_number":1,"created":"2020-05-27 21:48:46.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/56/731356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/56/731356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/56/731356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/56/731356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/56/731356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/56/731356/1"}}},"commit":{"parents":[{"commit":"c890e4314f8bd5d672e93fd890620e7802ea6b9a","subject":"Switch to Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c890e4314f8bd5d672e93fd890620e7802ea6b9a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-27 10:16:59.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-27 21:48:46.000000000","tz":0},"subject":"Consolidate python memcache bindings install","message":"Consolidate python memcache bindings install\n\nConsolidate all the installs of python memcache\nbindings to use ensure_package. Right now we are\nmixing the usage of ensure_resources on some places\nwhich causes redecalaration issues.\n\nSince the resources need to be identical in order\nto not raise redeclaration issues we cannot have\nthe horizon-package tag on the package.\n\nChange-Id: I0a76cbe45774bbf6871c1803c056bc2f6b3b6c0d\n(cherry picked from commit 3ec253827c10b81f61bf3b96cd85a50c2c264abc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9c8e18b5baba4c427eda4d8f505907a89ccd195d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9c8e18b5baba4c427eda4d8f505907a89ccd195d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
