)]}'
{"id":"openstack%2Fpuppet-horizon~731633","triplet_id":"openstack%2Fpuppet-horizon~master~I2e56b51f2ed3404c12f30b26c576e8a85a2b8ae6","project":"openstack/puppet-horizon","branch":"master","topic":"dnm","hashtags":[],"change_id":"I2e56b51f2ed3404c12f30b26c576e8a85a2b8ae6","subject":"DNM: Test tripleo gate job","status":"ABANDONED","created":"2020-05-29 00:55:00.000000000","updated":"2020-05-29 15:11:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7dbeb0788517c146e960764b04f65ad02d4b22bb","_number":731633,"virtual_id_number":731633,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-29 02:33:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 02:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91c2da2854677347e097058a918cefdbfdcb8cd4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-29 00:55:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"593b62e3bc2ce6c282bfbdf43c7072cbc87ad030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 02:33:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fe005b3c3b6475d8cf9296a6a181045 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bc0ca175c9441d083ece5844de145bd : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3b463114e91431ab76a9b67c3839de2 : SUCCESS in 5m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/83cdf55795694af4bef2c2f7cd4b9601 : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/155f4bb07b404b948c6b5c3be59a8697 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2fe80f9b26e4adc990f9ceed1c7797b : SUCCESS in 12m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31f9a854fda14dd4ae39ebafbac212fe : SUCCESS in 14m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22f545abfc4345d8aa2b400e7911ae85 : SUCCESS in 50m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4fe271ecf33844ca89e523efcdab8604 : SUCCESS in 1h 08m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/08cbd4ac0dfb45159946f8bd898e8259 : SUCCESS in 17m 05s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/61d72ecbda3946bfb0cafeca5d8977ae : FAILURE in 15m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd3ad97a2bdc4e6287944532b0ae0e56 : FAILURE in 9m 51s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/abbe95c796d4486fa93af384881ae319 : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8d3f6ec24a6e9ae37c696180f0f6e38115037210","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-29 15:11:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4728ec903341b751889482ad0a656d23bc85c34a","revisions":{"4728ec903341b751889482ad0a656d23bc85c34a":{"kind":"REWORK","_number":1,"created":"2020-05-29 00:55:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/731633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/33/731633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/731633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/731633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/731633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/33/731633/1"}}},"commit":{"parents":[{"commit":"482d061f9ccb71246a8681dedd4fd60aa7adbb1a","subject":"Remove useless parameters for Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/482d061f9ccb71246a8681dedd4fd60aa7adbb1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-29 00:54:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-29 00:54:44.000000000","tz":540},"subject":"DNM: Test tripleo gate job","message":"DNM: Test tripleo gate job\n\nChange-Id: I2e56b51f2ed3404c12f30b26c576e8a85a2b8ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4728ec903341b751889482ad0a656d23bc85c34a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4728ec903341b751889482ad0a656d23bc85c34a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
