)]}'
{"id":"openstack%2Fpuppet-horizon~758041","triplet_id":"openstack%2Fpuppet-horizon~master~I4d3b8623924ad646bf4880ad70fb3ce9be1675c2","project":"openstack/puppet-horizon","branch":"master","topic":"758041","hashtags":[],"change_id":"I4d3b8623924ad646bf4880ad70fb3ce9be1675c2","subject":"Rename horizon_* ssl params to ssl_*","status":"MERGED","created":"2020-10-14 09:03:04.000000000","updated":"2020-10-16 08:23:32.000000000","submitted":"2020-10-16 08:20:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758041-1602836458800-b16e9fe7","meta_rev_id":"6cd56641d326ae218586a8b977e961484191af2b","_number":758041,"virtual_id_number":758041,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-16 08:20:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-15 20:19:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-16 06:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-10-16 06:37:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 20:19:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-16 06:37:14.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-16 08:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31009981d9596c450e10cf8b8a10bebe8f1a4fbe","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-14 09:03:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b9892223b02f6a24202f8a3eb9e2503e97cc363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 10:42:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d6684fe08ef4dbe9f6b71dd932c7ff0 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/261835c4c0cb44c68f86d73b03d2b107 : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5bcd666499904ee3b4881a0d232f04f5 : FAILURE in 12m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5171284aa199497e9dd1b2f5ec2452ff : FAILURE in 9m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6373f9f45cc45d48e836adcca886441 : FAILURE in 11m 12s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/478dcbfe2f6c4d3da8ad50b16b35e872 : FAILURE in 27m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ad3849345b2c40b3aa0eb6d40ca1a868 : SUCCESS in 1h 29m 51s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ff31777b90374606845003c0fa4a15e4 : FAILURE in 8m 57s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddda6a4cceef4c6a8e1bfb9666f51c5a : FAILURE in 8m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6937bf498c8d4422ab0799650681348e : SUCCESS in 3m 36s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7078564ebf8345fdb59533038ab7bd1f : FAILURE in 57m 11s","accounts_in_message":[],"_revision_number":1},{"id":"693b99ff10c8a8d2cfdb17eabac0cba58cf7d0b9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-14 11:44:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2e1c3db43cf80db9e9619ce3cca965b5e666a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 13:39:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46056a5e9e2e417a8e358eb7da64640f : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b3e47e4f3934b7aa4ba92cbc83998f9 : SUCCESS in 3m 30s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/40081707d5284d65991f9eff6c893c85 : FAILURE in 12m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/774d63924eb649cc86314137cd298cd2 : FAILURE in 10m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b140a790ad9f447391ab6b18b16bd209 : FAILURE in 10m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11dde3b47103456a8e85ec68ef7cdd4d : FAILURE in 27m 48s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5fdd502b7ef2448d981e568521ef5bc9 : SUCCESS in 1h 35m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/697d15918df8456d990c91fd59a28ef2 : SUCCESS in 14m 07s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02bf94a0c5f342adb4a4fb0efa970139 : SUCCESS in 18m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/844c9591ba5c4d7d93f2785cdd22e049 : SUCCESS in 4m 07s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9758f06f11e4495f839328e657970e5c : SUCCESS in 1h 19m 24s","accounts_in_message":[],"_revision_number":2},{"id":"8ab5ff020748a3bac055d625fa91c235f46a47ff","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-14 13:47:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e8b96ac0a9545c8b15021ea1ae07033f589b801f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 15:41:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b507e08295c74123928da3a9a2a7b1fb : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5a32c302d174f93a41f9983c34d14fc : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7e65729f83fa4f10a48e9cd217b20cec : FAILURE in 12m 51s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9eb0dcd85c74194a6ac8e3aa9430874 : FAILURE in 10m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/786ba9b6d80f45489d67056b9530a2f7 : FAILURE in 10m 38s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32043a3874ed4a82a5f83d7fc1e5d006 : FAILURE in 24m 45s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/350f2dfc59344caea8bcdf1a242a32e7 : SUCCESS in 1h 27m 55s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/e8b243a2ab5348398d40ed8c32a96912 : SUCCESS in 14m 42s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/803ed5bb1e0c4a61b7f90d4d801ec22f : SUCCESS in 34m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6c992adf0a48cba345e6998b8aa51e : SUCCESS in 4m 17s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c34cd0a0cceb45c6a71f791a6cdd36b0 : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":3},{"id":"bd7d41cf27fbfa860a43666d5bb007dd020db626","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-14 15:56:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"22268c1b6f4629c0112bd1c629bef5a6e213d2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 17:39:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d45e530ffd04171b291833a918226f8 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3b88cc94a3b43ebb39a553fbc51be41 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a89ac1ae6b964285a529eed3aa64e9dc : FAILURE in 12m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef6ef4e1fd614dc98a89eb29cffd914a : FAILURE in 9m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7d3c1508aeb40208687ca50bc979c5a : FAILURE in 11m 34s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1b615e0933840289a3c2fe4b56f6004 : FAILURE in 24m 42s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/11faaba5beb04ff195bc39d88a06fba7 : SUCCESS in 1h 27m 27s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/de4c386cd4c744d5b0606e4024b111d1 : SUCCESS in 15m 32s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e95ad92e2114b50a745885d4d33300b : SUCCESS in 9m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e8083f9d625495cb5f5f1fdd40850c1 : SUCCESS in 4m 10s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c07453c5c149426b9ef7511de6b01a3d : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d35a962a62fb56d54767779ca4d76c2f8ef475f0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-15 07:48:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6f57d9edd940b433f2f21563e022b3768b0a4ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:24:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbf5ea1d9f334976bb8f76a62c11f9d8 : TIMED_OUT in 32m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e6185c091374f4e979b6f9c4ae9f225 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/32e25a43a4c74f3bacb87204e04e8ef6 : FAILURE in 13m 41s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8623f9a78da14a4d9226763724cf4190 : FAILURE in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d90bf50a6270494e9aeea19038f47c5f : FAILURE in 12m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35f95d2a945f45deb5496cd0534ebb78 : FAILURE in 23m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/78a5d3ed893b47629db7748061e72499 : SUCCESS in 1h 11m 20s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/27ac0d9b01004add8e07e2030698a332 : SUCCESS in 16m 26s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf870080ccda4df78aece563512549d8 : SUCCESS in 11m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7bcf691282a4e02a77387904a6e35a1 : SUCCESS in 4m 07s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/95b8ee68f8ce4125a1641238651813c2 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":5},{"id":"0d81378c0409282c4194ef43ff1e7ef7ee4d9c03","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-15 14:01:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"093d36ba2181d2845730cd463056c2f1dccfd6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 15:47:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb4f325062ae467385596e94fd67267d : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/446fe978941540eeb87da161750b6ab3 : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f5f4d01fd49e4a249ace57956b09eeda : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58192ea8bc9d4b77b030644e14a92a60 : SUCCESS in 10m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7897118c7e449029805e4fda6a71ee1 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2865c16bf8044aed9ea6289dd6445882 : FAILURE in 27m 25s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4d497d4047864b58b97a64aa7e7a836f : SUCCESS in 1h 38m 17s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9194330756914b7bad47aa24348a8a8e : SUCCESS in 13m 02s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f3139d181d6484ab13686b4261c6df5 : SUCCESS in 11m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0d8aea822f94eb8bfbf7c8db34f06c6 : SUCCESS in 4m 05s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee2ed9d75ccd4ee6924111a8032de019 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":6},{"id":"af649b78eb624cc8168e1d80f5e9321a106eea47","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-15 20:19:51.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"25eec51485f9d35d498836d80aad0a7f0ea8a1d0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-16 06:37:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"61b779ddef66e1f2dd0c0fb30b5f025e1a82751a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 06:37:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"44f09064e9e944914a52c027f953b557d2328dd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:20:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"39b8fce00b0abcf8be61f7a58504d3bf4c362a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:20:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f09bf4f1117473987da56e6abace54b : SUCCESS in 6m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/138774591f6746e183ef32456b1a8434 : SUCCESS in 6m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7b32848a99454cf19d2db34da0e23104 : SUCCESS in 15m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/316db045dd074c5591ae2d34cf3389f0 : SUCCESS in 11m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1acf105bfde54148a750dae8b9c4dd8b : SUCCESS in 1h 35m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/be62721f3f464ab29489af2473d604ea : SUCCESS in 18m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07a823539b7d41d4a5a0fbfd3e44da38 : SUCCESS in 14m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567e1e49450d4b9b807fdd284976beb6 : SUCCESS in 7m 12s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0f67658642584f3496067eed8effaa1e : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":6},{"id":"47054e033db2792ede840573148e3a6aa0841529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-16 08:23:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6647f5ab7b4da2b2e30cb388b3912c : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b2320ac2956ac699fa1caa71d3b13efa77cf7512","revisions":{"f254ebad9d56d04d153e8f3490731a6680fe3817":{"kind":"REWORK","_number":1,"created":"2020-10-14 09:03:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/1"}}},"commit":{"parents":[{"commit":"42ff35fcfbbcc1c3ba059a30ad8c95ff962ab634","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/42ff35fcfbbcc1c3ba059a30ad8c95ff962ab634"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f254ebad9d56d04d153e8f3490731a6680fe3817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f254ebad9d56d04d153e8f3490731a6680fe3817"}]},"branch":"refs/heads/master"},"4846a24ea8252c31e92e392e750e4a0b642230f9":{"kind":"REWORK","_number":2,"created":"2020-10-14 11:44:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/2"}}},"commit":{"parents":[{"commit":"42ff35fcfbbcc1c3ba059a30ad8c95ff962ab634","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/42ff35fcfbbcc1c3ba059a30ad8c95ff962ab634"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 11:44:27.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4846a24ea8252c31e92e392e750e4a0b642230f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4846a24ea8252c31e92e392e750e4a0b642230f9"}]},"branch":"refs/heads/master"},"332e17303bd8952b2bd00c34d78537ec445d3383":{"kind":"REWORK","_number":3,"created":"2020-10-14 13:47:17.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/3"}}},"commit":{"parents":[{"commit":"841d4d734e31f17cb1325d573df90c93d1f1683d","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/841d4d734e31f17cb1325d573df90c93d1f1683d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 13:47:06.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/332e17303bd8952b2bd00c34d78537ec445d3383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/332e17303bd8952b2bd00c34d78537ec445d3383"}]},"branch":"refs/heads/master"},"92cb75a05a466984b5693fe2fdeab64fed8720e4":{"kind":"REWORK","_number":4,"created":"2020-10-14 15:56:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/4"}}},"commit":{"parents":[{"commit":"841d4d734e31f17cb1325d573df90c93d1f1683d","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/841d4d734e31f17cb1325d573df90c93d1f1683d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 15:56:32.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/92cb75a05a466984b5693fe2fdeab64fed8720e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/92cb75a05a466984b5693fe2fdeab64fed8720e4"}]},"branch":"refs/heads/master"},"39bc8c0939462e647e95c6464443bd591e81614f":{"kind":"REWORK","_number":5,"created":"2020-10-15 07:48:50.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/5"}}},"commit":{"parents":[{"commit":"841d4d734e31f17cb1325d573df90c93d1f1683d","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/841d4d734e31f17cb1325d573df90c93d1f1683d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-15 07:48:42.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/39bc8c0939462e647e95c6464443bd591e81614f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/39bc8c0939462e647e95c6464443bd591e81614f"}]},"branch":"refs/heads/master"},"b2320ac2956ac699fa1caa71d3b13efa77cf7512":{"kind":"REWORK","_number":6,"created":"2020-10-15 14:01:00.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/41/758041/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/41/758041/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/41/758041/6"}}},"commit":{"parents":[{"commit":"841d4d734e31f17cb1325d573df90c93d1f1683d","subject":"Add ssl_verify_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/841d4d734e31f17cb1325d573df90c93d1f1683d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-14 09:02:01.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-10-15 14:00:52.000000000","tz":120},"subject":"Rename horizon_* ssl params to ssl_*","message":"Rename horizon_* ssl params to ssl_*\n\nThis renames the horizon_* ssl parameters to\nbe named ssl_*. It also adds an enforcement to\nthe end users to set ssl_verify_client if\nthe ssl_ca (or deprecated horizon_ca) is used\nsince it wont take effect.\n\nSee the puppetlabs-apache 2.1.0 release notes [1].\n\n[1]\nhttps://github.com/puppetlabs/puppetlabs-apache/blob/2.1.0/CHANGELOG.md#changed\n\nChange-Id: I4d3b8623924ad646bf4880ad70fb3ce9be1675c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/b2320ac2956ac699fa1caa71d3b13efa77cf7512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/b2320ac2956ac699fa1caa71d3b13efa77cf7512"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
