)]}'
{"id":"openstack%2Fpuppet-horizon~796019","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fwallaby~I1d97225939ccf9b8d0403caaa16ea3ebaf546c51","project":"openstack/puppet-horizon","branch":"stable/wallaby","topic":"puppet-horizon-dashboards","hashtags":[],"change_id":"I1d97225939ccf9b8d0403caaa16ea3ebaf546c51","subject":"DNM: Testing (2)","status":"ABANDONED","created":"2021-06-11 14:09:31.000000000","updated":"2021-06-26 17:55:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7296907eaecb85c22325bd10b66a56d36ae42b55","_number":796019,"virtual_id_number":796019,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-11 16:15:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-11 16:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa64c6dff7dc273be7f0712188d4c8a4e3ac0f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-11 14:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5025e1cb1cb0546793481f964669f80aad24ec3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 16:15:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8323d9cdb464540a694b4ea1ede99f6 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1270081b14d419fbca7862260fc4374 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/480985a1a46c49368a51f69dc7be6e29 : FAILURE in 15m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3c4971d031b46929d9593198f588256 : FAILURE in 14m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9546616361c64fb089a375dc157ea2bb : FAILURE in 11m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/956d14be1689466a90f5d4e3b9676d40 : FAILURE in 32m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee378d56712f41689088f0b7e3a20421 : SUCCESS in 1h 20m 53s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/8be858213f5545e8938bddc1be8e7747 : SUCCESS in 18m 05s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c73de5720974bdbbdc32914ae71c161 : SUCCESS in 11m 16s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4faaebd78d804fa8b50c080936333949 : SUCCESS in 1h 51m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7296907eaecb85c22325bd10b66a56d36ae42b55","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-26 17:55:54.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cd8e8c6134ea1b9af0835eefe10579fbb1fa708","revisions":{"9cd8e8c6134ea1b9af0835eefe10579fbb1fa708":{"kind":"REWORK","_number":1,"created":"2021-06-11 14:09:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/796019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/19/796019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/19/796019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/19/796019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/19/796019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/19/796019/1"}}},"commit":{"parents":[{"commit":"655e0df25e40b401d2daa5d673a416215ccd17b7","subject":"Add support for independent heat-dashboard configurations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/655e0df25e40b401d2daa5d673a416215ccd17b7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-11 14:09:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-11 14:09:19.000000000","tz":540},"subject":"DNM: Testing (2)","message":"DNM: Testing (2)\n\nChange-Id: I1d97225939ccf9b8d0403caaa16ea3ebaf546c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9cd8e8c6134ea1b9af0835eefe10579fbb1fa708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9cd8e8c6134ea1b9af0835eefe10579fbb1fa708"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
