)]}'
{"id":"openstack%2Fpuppet-horizon~802202","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fussuri~I5999c2eaf8d7788d3c0d76e69a56a0b10058dfe9","project":"openstack/puppet-horizon","branch":"stable/ussuri","topic":"puppet-horizon-dashboards-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I5999c2eaf8d7788d3c0d76e69a56a0b10058dfe9","subject":"Add support for independent heat-dashboard configurations","status":"MERGED","created":"2021-07-25 23:46:29.000000000","updated":"2021-08-24 09:47:22.000000000","submitted":"2021-08-24 09:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802202-puppet-horizon-dashboards-stable/victoria-stable/ussuri","cherry_pick_of_change":799983,"cherry_pick_of_patch_set":1,"meta_rev_id":"ca436847b28d7b8492493890e67a4572b861ea03","_number":802202,"virtual_id_number":802202,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 09:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-08-09 14:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-08-24 07:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-08-24 07:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 01:27:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 16:13:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2021-08-09 14:24:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-08-24 07:24:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0baa579169e068d53a972fbb8dafbdbbd20b37f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-25 23:46:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"bbfe5c38d9c67a91049288683ac30ca1c7090947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 01:27:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f1a29131efd464e9c025646324be19a : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5acd3f2997064aa680eb803ccdd8f6ce : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3584c3564dc242daa7707c0e029b2859 : SUCCESS in 3m 45s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1fba843515214cfda466197ebf7875dd : FAILURE in 14m 30s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/53ca801c9d80480da5a2ddbc9592244a : FAILURE in 16m 12s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3eee58003fdc426898bead178b6d42f0 : FAILURE in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/363a4e3b6b244edfa52b38d79d30750e : FAILURE in 12m 00s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00760c68f2b341bdaf2b23bb60aa8574 : SUCCESS in 47m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/81c0205296d14e8a89af39619cd4faa9 : SUCCESS in 1h 12m 52s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8e19f7e0cb14ae9bc7f8a6a439ce339 : SUCCESS in 1h 17m 41s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/798df9e488254ee4876a46cf8aab726c : FAILURE in 17m 11s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2f1f885eb68417aaa5e7bdff3eaff5e : FAILURE in 20m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce3ed5dc4b60419ea3a9fa165913e799 : SUCCESS in 10m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6157f65acb5e4cf9b86728ffabab6521 : SUCCESS in 3m 54s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3838ddb2fff14099b82c41705b632725 : SUCCESS in 1h 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1bc616b2b6208b91cbe253a7238b21e2b86ce96c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-27 16:13:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc5ccae502ac80648dea96bfc23ac9de76c26853","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-05 09:51:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"036ffe81287c6f553d34419f57a0032eaabdbb5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-05 09:59:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3de1a1c926c9294045e7677b30c6168ce7dab5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 11:45:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d0dc4f9391d4d118ae26804155f9f4f : SUCCESS in 3m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43b5cfd5956947ab81375a3789d66f66 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23de0afa08d64dff83e015cbeae8f0c6 : SUCCESS in 3m 32s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7154a11dc2034943a09c4738b42b600b : SUCCESS in 14m 51s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/404422ab53ec4779a21cdaaaeb0c71ae : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f9631a45e94483e9fd479390229ab64 : SUCCESS in 10m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/731627628d3e4ffdb79da90f2964e247 : SUCCESS in 10m 27s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93325d7454404d228ad1c816d32b82b2 : SUCCESS in 43m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9742e2085190432db7b01f19ce6cbf39 : SUCCESS in 1h 09m 30s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7d8461114e5495480045889e649f360 : SUCCESS in 1h 13m 43s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/eb2ce2cf74bc4a66a620e1467f64ca23 : FAILURE in 17m 33s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3fc4142254542e0ad4ee621946af39c : FAILURE in 19m 34s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f059beb3fbb463292216536375d3b4f : SUCCESS in 10m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6fb0110fd4f4352b8fb77c1f773c18e : SUCCESS in 4m 02s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4c094d9fe99e40b7b00c0e7f120428b7 : SUCCESS in 1h 43m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8e60871904b739b137ff4ba82a6e9bd81853bac8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-09 14:24:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9dd2a55ab73c50fe9dff9479c439fac461f25d10","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-24 07:24:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"316a14f6b848102bffeedabe7149697ac6652583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 07:25:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"418cdfe8f06167219c67903f56ea78f205dcd55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 09:44:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e7f5d707ff842a0a782d586e309601a : SUCCESS in 5m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41c032df150c4d78a487adec5861d260 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdd0a825af9a4ec384c52e9fe995c36f : SUCCESS in 5m 50s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/71ff262bb8474eb59fc4eb9bcb85b616 : SUCCESS in 19m 23s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef242d1882cc40e99fbf658819472235 : SUCCESS in 13m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed6da50ee8fa44b4b57702c7c22b6f49 : SUCCESS in 10m 59s\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fbf5cda6796247339fa7f4ea2cd02843 : SUCCESS in 1h 04m 52s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cb0c74b6460149b8858901bb9a6d63e7 : FAILURE in 18m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f256a34ec604345adc578185562e38d : SUCCESS in 4m 44s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/afd4828a00b4417299abb0063bd48f68 : SUCCESS in 2h 12m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5caf9eb0a51bcdee4b394525818a78e60a3194a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 09:44:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ca436847b28d7b8492493890e67a4572b861ea03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 09:47:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2a52d3cfcee4747a20c7ae488d73b21 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf021098cc262097fd7945f284d04b46c69a2b66","revisions":{"d346bbb2b156bcdd8e4d4f830cbfadfa82fbba30":{"kind":"REWORK","_number":1,"created":"2021-07-25 23:46:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/802202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/802202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/1"}}},"commit":{"parents":[{"commit":"ef402860ea14433b677a0bd5c2a6f2893f2f3f19","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ef402860ea14433b677a0bd5c2a6f2893f2f3f19"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-12 07:28:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-25 23:46:29.000000000","tz":0},"subject":"Add support for independent heat-dashboard configurations","message":"Add support for independent heat-dashboard configurations\n\nCurrently horizon supports per-dashboard configurations in\n/etc/openstack-dashboard/local_settings.d. Each dashboard plugin\nprovides a specific file to define its parameters, and any parameters\nused in dashboard plugins have been removed from the base\nlocal_settings(.py) in horizon repo.\n\nTo follow how each plugin is supposed to have its configurations\nmanaged, this change introduces an independent file for heat-dashboard.\nSupport for the other dashboards like manila-ui, octavia-ui, and etc\nwill be added later.\n\nConflicts:\n\tmanifests/init.pp\n\tmanifests/params.pp\n\tspec/classes/horizon_init_spec.rb\n\ttemplates/local_settings.py.erb\n\nBackport note:\nTo minimize effect of this backport, the following change was made.\n- The enable_user_pass parameter is not deprecated by this patch.\n\n- Users can set the enable_user_pass parameter to undef to remove\n  OPENSTACK_HEAT_STACK from local_settings.\n\n- heatclient logging setting is left in local_settings, because\n  consistent configuration is added.\n\nChange-Id: I5999c2eaf8d7788d3c0d76e69a56a0b10058dfe9\n(cherry picked from commit 382877dd4c8ee321eff1846c7c7afe4104dbd58f)\n(cherry picked from commit 655e0df25e40b401d2daa5d673a416215ccd17b7)\n(cherry picked from commit 80acdcca4dcbd058d72b5099054cc6b1b8b798b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d346bbb2b156bcdd8e4d4f830cbfadfa82fbba30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d346bbb2b156bcdd8e4d4f830cbfadfa82fbba30"}]},"branch":"refs/heads/stable/ussuri"},"bf021098cc262097fd7945f284d04b46c69a2b66":{"kind":"REWORK","_number":2,"created":"2021-08-05 09:59:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/802202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/802202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/802202/2"}}},"commit":{"parents":[{"commit":"ef402860ea14433b677a0bd5c2a6f2893f2f3f19","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ef402860ea14433b677a0bd5c2a6f2893f2f3f19"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-12 07:28:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-05 09:56:35.000000000","tz":540},"subject":"Add support for independent heat-dashboard configurations","message":"Add support for independent heat-dashboard configurations\n\nCurrently horizon supports per-dashboard configurations in\n/etc/openstack-dashboard/local_settings.d. Each dashboard plugin\nprovides a specific file to define its parameters, and any parameters\nused in dashboard plugins have been removed from the base\nlocal_settings(.py) in horizon repo.\n\nTo follow how each plugin is supposed to have its configurations\nmanaged, this change introduces an independent file for heat-dashboard.\nSupport for the other dashboards like manila-ui, octavia-ui, and etc\nwill be added later.\n\nConflicts:\n\tmanifests/init.pp\n\tmanifests/params.pp\n\tspec/classes/horizon_init_spec.rb\n\ttemplates/local_settings.py.erb\n\nBackport note:\nTo minimize effect of this backport, the following change was made.\n- The enable_user_pass parameter is not deprecated by this patch.\n\n- Users can set the enable_user_pass parameter to undef to remove\n  OPENSTACK_HEAT_STACK from local_settings.\n\n- heatclient logging setting is left in local_settings, because\n  consistent configuration is added.\n\nBackport note to Ussuri:\n- The log_handlers option is not available in Ussuri so it is replaced\n  by the old log_handler option.\n- This change includes what was fixed by follow-up commit[1] in\n  Victoria.\n  [1] 23ebe3bf02f903abbba885f17dc56590fcf6a51e\n\nChange-Id: I5999c2eaf8d7788d3c0d76e69a56a0b10058dfe9\n(cherry picked from commit 382877dd4c8ee321eff1846c7c7afe4104dbd58f)\n(cherry picked from commit 655e0df25e40b401d2daa5d673a416215ccd17b7)\n(cherry picked from commit 80acdcca4dcbd058d72b5099054cc6b1b8b798b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bf021098cc262097fd7945f284d04b46c69a2b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bf021098cc262097fd7945f284d04b46c69a2b66"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
