)]}'
{"id":"openstack%2Fpuppet-horizon~823932","triplet_id":"openstack%2Fpuppet-horizon~master~I1d77d7076f845e2fc47a4132eabe442091a63374","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-15 12:15:38.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I1d77d7076f845e2fc47a4132eabe442091a63374","subject":"Add support for DEFAULT_BOOT_SOURCE","status":"MERGED","created":"2022-01-10 05:21:16.000000000","updated":"2022-01-15 16:49:46.000000000","submitted":"2022-01-15 16:48:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823932","meta_rev_id":"b657849e8533c349521b9beea10e4f1142708f14","_number":823932,"virtual_id_number":823932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-15 16:48:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-10 18:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-13 08:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-13 08:58:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 07:08:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 18:43:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-13 08:58:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"4b52324ad3dfe29d70b0f51b5c97412288e7544f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 05:21:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37aef7f0e9d6581717081467a8564428879f3479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 07:08:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/100e734b89b54a0783ee2e8b26238cf9 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e855ba6d590e4b9dbf51fcb718a0a778 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc5b927252dd40de8682673fd2a4f738 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b697ac4405504c44837f9fec999cb495 : SUCCESS in 16m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a046629a9ddc4d7d914f0686a6995a84 : SUCCESS in 15m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f6f1924dde46a2ae2b768bb2d6c803 : SUCCESS in 13m 37s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b22198a8b5f41eda5a8e791e61b4191 : SUCCESS in 12m 30s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0e7b78cf840448f82e75bb9c030540d : SUCCESS in 12m 27s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5549df409a8f42da9a6741c5d4c958ee : FAILURE in 21m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/14eef6aa6ce144a9b7ca2a795f756e85 : SUCCESS in 1h 03m 20s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ecd1edd9503498a9d831d84672e6b5c : SUCCESS in 55m 49s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6894a2eb05a448d48779e629ac6d4143 : SUCCESS in 11m 34s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/480e576e8f9343089d1f8edc45eebc7e : SUCCESS in 20m 31s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71ef46515c5848ef896f176626b1b214 : SUCCESS in 14m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5741b9c64c6e4167ae5d30f59de9e4fc : SUCCESS in 12m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f38563b49574df0b6057bf7631cca54 : SUCCESS in 4m 48s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/80b01fbe627e4174babf1d7d7b9ca124 : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":1},{"id":"35a3c2d45355b06bb26a74f70ab4e372c69d93e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 07:35:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d7aa37248df4feff3f5ece43b2410e241de8fa55","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 07:36:07.000000000","message":"Topic bug/1956904 removed","accounts_in_message":[],"_revision_number":2},{"id":"3c51a5a25123b4beab6d46921e620489a87aabf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:43:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f63bb7bf7bf64582a6f2695d384609f0 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64127852dade476c8019536efe4d3ef3 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603bdcf8ab5548ab839d290eca34619f : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d04691af7e8e4561b44aa14dbdcf5a77 : SUCCESS in 14m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ccf4fbe27b6420e81b5fdf3543d51f3 : SUCCESS in 13m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d95a0560b5246da83ab77bfe935a6b2 : SUCCESS in 11m 46s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eddadcd8870842bf8d20ac9dc1faa313 : SUCCESS in 13m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/212712df93c74099a8b982a33e4249a0 : SUCCESS in 11m 35s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f116e85b5e014d9f967b2b36cd4b2271 : FAILURE in 43m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/930f34e0c3a942919acca349290458cf : SUCCESS in 1h 25m 36s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/47ccf23ad2c04640bd893614c4ced0f7 : SUCCESS in 1h 03m 46s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca15501d6ae4a0e81b22e4525c151d7 : SUCCESS in 11m 10s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9cfcd5da634345a68a1066aeaced4773 : SUCCESS in 16m 50s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e345dc02375147bb88dfc27807b05ef5 : SUCCESS in 10m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/756bbfba1f5043dc85d5496d01d41ba1 : SUCCESS in 9m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bcdf8bafed34692b7f87bced3e53888 : SUCCESS in 4m 19s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5504f8ed09544db1b8871af916d0ccda : SUCCESS in 2h 05m 42s","accounts_in_message":[],"_revision_number":2},{"id":"536434e25167fe4a73799184bfe3abedebec8c72","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-10 18:43:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e198082846d5786ebb3651e3aa166b65aea46e7e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-13 08:58:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"eaa8fa9be97a01a8b076977837c567de90250e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:59:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30ee34e086a6221148867dbf3c044f3888f06023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:21:13.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f26dc93feb1ca4b9d90312d36c7f8b885e11b000","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-13 13:31:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a806b5ca5b4a81b380d9700a5bc4fe0f6ac4318a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 16:23:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dce075eabfe64c018cb1a1a3bf5c7760 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02bcf23025284f5aaf0f3cdd2f0288f1 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7650296080b74cf69734557f2900234b : SUCCESS in 4m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc6f86a843ed407ca63fa2e97b139064 : SUCCESS in 21m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c39abf7818914aad82ef9ceed4fb794f : SUCCESS in 21m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b2795a5d844379a3cbfd288d8fdd6b : SUCCESS in 18m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73a0ad2f0efe4ceb8e4b7b260e1342e9 : SUCCESS in 17m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca454bf7331145b39348e5a20115ea2b : SUCCESS in 22m 05s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be50388027d6461e9589509e7d97ad1c : FAILURE in 41m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f69b9ab3cfa8458ebaa875e478a6f5ac : FAILURE in 1h 41m 22s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f59a2e23ed940cfb494721ffc83c535 : SUCCESS in 1h 12m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f0121ada284476ba6e56a95bb86c12d : SUCCESS in 16m 58s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f941c0de59594b089bd0ba129f91799e : SUCCESS in 24m 30s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc8a801b5b64e9eb36581d12d43048b : SUCCESS in 14m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0017d8ffd98f419882c8c5167185f9af : SUCCESS in 17m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bb09a67c1df42d287679faa5787e24f : SUCCESS in 6m 09s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fd30cab4c9824bde97bd1aa2bb3edc5a : SUCCESS in 2h 14m 11s","accounts_in_message":[],"_revision_number":2},{"id":"dd4896e7f990bf405e084e6aa115ab841147a29a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-15 12:15:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aae1432e23d73ff6c2395e3b27045aa5c18c02ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 14:51:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6807816d3f63439ab8b50ce63e1f0317 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/916e85e616334295ae02fab657a63825 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ed8ed17e6284843afdb7d697811b875 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa8b9d8626594dad83fae51de805b5bb : SUCCESS in 13m 59s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ec92dad9c83400890bc5edd1f31320a : SUCCESS in 16m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20e9ddef76034c71b5ce2de03baa0063 : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44fd1db8295d47a59248ba04851253d4 : SUCCESS in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ddfbc6430be40c0bf5add942fc59bc7 : SUCCESS in 15m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf5ce77547a424498f46d7891e5d367 : FAILURE in 34m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce32800dd3554013b248a3e5fb9d83c5 : SUCCESS in 1h 31m 24s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/889aa48abf7c4691ab4f11f790c6878d : RETRY_LIMIT in 5m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c958176ce6e143758a4e21525fd5eead : SUCCESS in 12m 20s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1c0ea8d452a4fd28b158453b340121e : SUCCESS in 18m 25s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2f1b88b7cbe44c5848701ea98f5ec2b : SUCCESS in 10m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a51fddcff7164e9fbe35573b9eed8190 : SUCCESS in 11m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0790e20285bf47d8919f5b2748b4013b : SUCCESS in 3m 43s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9caa39cdc9be4c2abb1a19b159da5a18 : SUCCESS in 2h 27m 44s","accounts_in_message":[],"_revision_number":2},{"id":"01bb5891f88cef606c10bad826168caf4642f91b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 14:53:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8813c8dde786c3aed00453261a0d06a6dd2431a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:48:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547ccd9c73cb4c7fb7a1bee7307cf442 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8283da5dfe07414e8aaa5536b5b5c73c : SUCCESS in 3m 28s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30fc75e149fd4e52aaf0d986a6ae6aa0 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9602a8235f234344a6bbde363ba60282 : SUCCESS in 13m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2515bf518d44d3fba4e75e2ce940972 : SUCCESS in 13m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b3a35868d424cb3a955d51918bb2bd8 : SUCCESS in 11m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4406a8901f8647bc92f56e85f9d9b545 : SUCCESS in 11m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/18f3674131a049b9b278f98a93bc234e : SUCCESS in 1h 12m 01s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6138c46c3689457eb558e8384b8c5dd2 : SUCCESS in 15m 15s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d1d11c9dc734419aa3245166f9d9821 : SUCCESS in 10m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93dca061246f4c0591c7f99a01ea33a8 : SUCCESS in 3m 33s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3e629bc1a602434681ba22f34b16b596 : SUCCESS in 1h 54m 07s","accounts_in_message":[],"_revision_number":2},{"id":"427cb1c632cb040bbd4be1ed0929367158072d4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:48:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b657849e8533c349521b9beea10e4f1142708f14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:49:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8d1aa315c84a0f95c586b209f964c5 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32f52c19ca2221db1c6bd46f260005540069877b","revisions":{"49b79657b4d3498d8179a4971138b6f964974b7a":{"kind":"REWORK","_number":1,"created":"2022-01-10 05:21:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/823932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/32/823932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/1"}}},"commit":{"parents":[{"commit":"518976f920a41c4546608c6365d6428722080fc1","subject":"Add parameter description of HORIZON_IMAGES_UPLOAD_MODE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/518976f920a41c4546608c6365d6428722080fc1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:21:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:21:01.000000000","tz":540},"subject":"Add support for DEFAULT_BOOT_SOURCE","message":"Add support for DEFAULT_BOOT_SOURCE\n\nChange-Id: I1d77d7076f845e2fc47a4132eabe442091a63374\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/49b79657b4d3498d8179a4971138b6f964974b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/49b79657b4d3498d8179a4971138b6f964974b7a"}]},"branch":"refs/heads/master"},"32f52c19ca2221db1c6bd46f260005540069877b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-01-10 07:35:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/823932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/32/823932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/32/823932/2"}}},"commit":{"parents":[{"commit":"e7590f702653f88281caeeac6f0b2ba2aab5b4e4","subject":"Add parameter description of HORIZON_IMAGES_UPLOAD_MODE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e7590f702653f88281caeeac6f0b2ba2aab5b4e4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:21:01.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:35:33.000000000","tz":540},"subject":"Add support for DEFAULT_BOOT_SOURCE","message":"Add support for DEFAULT_BOOT_SOURCE\n\nChange-Id: I1d77d7076f845e2fc47a4132eabe442091a63374\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/32f52c19ca2221db1c6bd46f260005540069877b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/32f52c19ca2221db1c6bd46f260005540069877b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
