)]}'
{"id":"openstack%2Fpuppet-horizon~823933","triplet_id":"openstack%2Fpuppet-horizon~master~Ie3c13d96e8219b4d80b0641f4def49b858f0513a","project":"openstack/puppet-horizon","branch":"master","topic":"bug/1956904","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-01-15 16:33:05.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-15 12:15:31.000000000","reason":"Takashi Kajinami replied on the change"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-01-13 08:57:46.000000000","reason":"Tobias Urdin replied on the change"}},"hashtags":[],"change_id":"Ie3c13d96e8219b4d80b0641f4def49b858f0513a","subject":"Do not hard-code REST_API_REQUIRED_SETTINGS","status":"MERGED","created":"2022-01-10 05:23:01.000000000","updated":"2022-01-15 16:34:17.000000000","submitted":"2022-01-15 16:33:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823933-bug/1956904","meta_rev_id":"7a9e585441e54ed9f6b5edf20a37a9c69939a688","_number":823933,"virtual_id_number":823933,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-15 16:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-10 18:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-13 08:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-13 08:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 09:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 18:42:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-12 04:44:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2022-01-12 04:44:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"44dd9b04d252a588cd77b19deccf5ff76823b806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 05:23:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1a822ccd89a92e3f8a636f51e4b136853b7c874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 05:31:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"014411023efede575a183dd090ecc1d42be93384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 05:36:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09c869be9f3a6b00519189f250a48a8f6923ae31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 07:34:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"61c9de97a0fc25116e7ea0a3e54f4cdfe6058798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 09:44:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6fde061d1484b6ea4c30326bd10ec0f : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dabcffcba024a9795adc5d7ca21ad3c : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c46dcb3c0cc46239d7e75a859d8c821 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b050b49e5e924c5285b2e85ee4c849dd : SUCCESS in 16m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/58b53396b6ae4aaebb58fefa011279dc : SUCCESS in 13m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/624f141d52da4c4480d6df6da3d77924 : SUCCESS in 13m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8d8cd8d28ca433db06108e78a2ecad5 : SUCCESS in 14m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7bf37165cac4ac5bc066d3c67f86ffb : SUCCESS in 11m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/560c622eb0744001ad5b39072526e624 : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bac7da0ddf94beb8a24e4a0d430f11a : SUCCESS in 1h 25m 30s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e814f6f0302404e867923c03b68f87a : SUCCESS in 1h 02m 58s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/038b9387d275432ea659e4ba923abf4d : SUCCESS in 10m 59s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c35dc537875449a963027633bda1dd3 : SUCCESS in 17m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b760e9bad9443beb0513d88b02763fd : SUCCESS in 9m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a0b4d37e477442d9a9a5648b50c6223 : SUCCESS in 9m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/287db3b1e5874630ad3eae19be33afa8 : SUCCESS in 4m 26s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/04c706fa740a408d85bc2b2df3c0b891 : SUCCESS in 2h 08m 47s","accounts_in_message":[],"_revision_number":4},{"id":"61d9de24bbc2cb53479652595ee87e87481951d0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-10 18:42:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"6b07d8d9d31ef5b3e5867f69585fd137138a124a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-13 08:57:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ec663442bb4df507b4f38b5498d0788321d94c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 08:59:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d753e596b9fff4b075558058caf8de67b8d26427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:21:11.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9364b415c77d400491b4f951e9225859 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/761f484bca5e487ab75bd1c1b6101b1e : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de63e02a90c045e280208308161174ba : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/30fb97b6953f463ba5b7099e472b01ee : SUCCESS in 17m 56s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a39abfdcf2124349bdf61d2976f82399 : SUCCESS in 18m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa819e4f53f4092ac2b36c264ba0af1 : SUCCESS in 15m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39612571f44946e2aa1389f3f082a03f : SUCCESS in 14m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed013cb4ea9542e7a50bd1a6a63af98a : FAILURE in 1h 19m 54s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef4d97726d30464e9d2d2fb02f22eaeb : SUCCESS in 20m 34s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe65b322a4224554b9312da79181d41a : SUCCESS in 13m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c73ec6d12c1447bc8f53e7d5c72c1dcb : SUCCESS in 4m 52s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e809218a6b6c4272b98cbf9b067704b0 : SUCCESS in 2h 16m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6dadafd3b415ce6feb9dc7f331a0d4ebad153292","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-13 13:31:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ecfa2c4769a9231b4bbefc77ed4df750442ab955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 16:19:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec45fddd86764d799beff9635ddf8765 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e0082852f584904804a10aee7117102 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bb2724a3d6f496ab9333a8ec63b24a6 : SUCCESS in 4m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c430d71c163423fb7f99910752b6866 : SUCCESS in 17m 18s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9237a3e5bc541d695ced471a486a195 : SUCCESS in 21m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204f907978244fdc84db178dc8c0bed4 : SUCCESS in 18m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe412ccd94b04419a82ae1a5a15d70ab : SUCCESS in 19m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/426b979033a44ccba81a57a0e48aa95b : SUCCESS in 15m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd54073e0453492186ed48a3d9260174 : FAILURE in 39m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9037c9ce1c18409988c79061a71d0ef8 : FAILURE in 1h 27m 27s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d7995fbe169489fb5e050df399cdac0 : SUCCESS in 1h 15m 37s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c22e5f5db1ec430e8396620884a587fd : SUCCESS in 14m 57s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/efa0b6676d93474ea65c99ee1599048e : SUCCESS in 22m 40s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25294bbde7d34772a54aed6a5cf65b0b : SUCCESS in 15m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c7694ba2b094a4ea95250b86b04f1fc : SUCCESS in 12m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31a7200960f34566aeadf4de100057cf : SUCCESS in 4m 18s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d8c9f73eaa54890931e09183e0e8b32 : SUCCESS in 2h 18m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d2a5973eb7ca0aacb8d143bbd9c522de60fea1ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-13 16:25:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04846c9551e983a8940cbedf869fc3a3bf569401","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 18:33:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04f533df88984708af6fe097f1655bea : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9ae201c28994cb89a735521c0570269 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/515caf256576437cb0569fa85a31a273 : SUCCESS in 5m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a938598a4304aa384d0ed81922f0697 : SUCCESS in 16m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/719ab370597d4a578cac9177109a53f6 : SUCCESS in 16m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9bc05dd86c040caa05ef6daeb6ad9ba : SUCCESS in 15m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b25f80158584443c831fdf2ccef0e651 : SUCCESS in 14m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed93c2bc466048018bc3e8e06210c7ca : SUCCESS in 15m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3b680d0bea48c2888357d7896c03f9 : FAILURE in 34m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/411f976f07264bf882a0cfe61413a736 : FAILURE in 1h 43m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e64b83721c944b7a94200c65d5a7c7fe : SUCCESS in 1h 01m 34s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeb1b948faae44ec885ed5fdf8ffe845 : SUCCESS in 12m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0207356d24d24992a0552d4d99fb8de6 : SUCCESS in 21m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0746c544581849c0836781befb8a2fd1 : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/492c41793a484c6f9fa8b6528ff63364 : SUCCESS in 13m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b28654e52394648b3ebc48f21cb7283 : SUCCESS in 4m 21s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2f9185e8ee43411d84e99d8a997be96f : SUCCESS in 1h 59m 55s","accounts_in_message":[],"_revision_number":4},{"id":"10b40133a2eb16ab304b783955d401aea73fa2e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-14 00:55:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"51a28428774c9cc56d0a1357bd5656cf824f1085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 03:03:00.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e2a441790324ec59c59508fa9d3de61 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7741f6e2cdd844ca92b4031ba211c911 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a5515705e6140e7a0e10b99b3c4cc60 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/44cf6295e9e744669a9e5ea7faf4d892 : SUCCESS in 17m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/08da7d4a0ee7451cb8534b683f6f9ed0 : SUCCESS in 15m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e01bd04f5a246ab9f4177c29f982937 : SUCCESS in 16m 13s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77dffc07627f4feda9c62af9aeee09ff : SUCCESS in 15m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ad528062ab46deb1c33cc7773120b2 : SUCCESS in 11m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02b4b6ef98f74fbdb489a3127e0edd9d : FAILURE in 31m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d697ae697524c5a8a346d122139802a : FAILURE in 1h 20m 44s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d8bd65428c9420b8d3b059851f90dcf : SUCCESS in 1h 01m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51c0bafbb6fe42cf8c22cd31f394e6da : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/83b32d11e73c48acbcc118a4ecf3c1d7 : SUCCESS in 15m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d712d9bb394e39ab08168eaeb8e040 : SUCCESS in 13m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d704a36218ca4896b41d0345f17da10f : SUCCESS in 9m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d56c3ae606904b0fbeee1384e305a80c : SUCCESS in 5m 48s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f70690f0b8ec4134b7f9fb917f8f5216 : SUCCESS in 2h 04m 15s","accounts_in_message":[],"_revision_number":4},{"id":"cbc35f2e7ae368cf4b5864d2f52431b893561565","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-15 12:15:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b6b6b017c706e1d7e674aeaaa7621c79dd871680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 14:34:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158dd1237eee4a039b446885155b1f42 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bae5faa56574f1ca30526c2f1edaf34 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c624bac214724fabb8b235b96fbc0c8c : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3412969ecdd5454faa69b137844d58d0 : SUCCESS in 16m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2375a5f095ae42c698ea37c68ffe2856 : SUCCESS in 17m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/011ce4e8855d4ebda2f6306da6223f8c : SUCCESS in 11m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca2c8932b628419899ab69af5ea1a3e3 : SUCCESS in 14m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6b0884268f046a19c7aa23672194049 : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c7d4f14538f4e0e940343ec828f9389 : FAILURE in 27m 55s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b77d92f2afa4486a9bb11ad2db628d71 : SUCCESS in 1h 29m 53s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/92611c580507408e866269e73904bb4e : RETRY_LIMIT in 8m 20s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/648c8f1b90994665b823cd171d63b226 : SUCCESS in 10m 20s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/97821dc3ce724d14be3eab076ebad1c3 : SUCCESS in 14m 52s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cbec8876d4a46ffbe74fd699cfd53d6 : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a51817ebb6084c47858056f445e6eac3 : SUCCESS in 9m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1696f4d8fe6461f898fdbd3a6b1c1df : SUCCESS in 4m 13s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d71cbf9dcde844699ccbc9fce76a1cd4 : SUCCESS in 2h 12m 38s","accounts_in_message":[],"_revision_number":4},{"id":"eb83a32e0feaedb09971f3d65c90bc346bda87d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 14:35:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f78703ca696ada0956e0601f8633aadb4d3681d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:33:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fef98807174aee9d2f867b72d2b3bc : SUCCESS in 2m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5caf9ddd7f7149b189e97a9fe1cb25ad : SUCCESS in 5m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c7aa5627c664a2685a9abb8eaee2e48 : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b6a67b5a7634c5fa8f4e847c494f274 : SUCCESS in 11m 10s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/042a13e719564755842ea30ab05d8f40 : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff07d4bd1b0497bb52d7204d9363ae2 : SUCCESS in 9m 45s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e788f147625c4db7a910b9635b54fc29 : SUCCESS in 9m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/91c5913c972b406e9c3ec2643331e173 : SUCCESS in 1h 23m 03s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/996133e133bd4c18b7b8a2279aa8f85e : SUCCESS in 12m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f01b263eda854bc09328740758d00a63 : SUCCESS in 8m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf915ef88e94abfabbf5ab1f870fe88 : SUCCESS in 3m 17s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/52d67475bf554448ba7ce6dec04b6b87 : SUCCESS in 1h 56m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ed703f8544afc7ce5f4c58d3ef37310750eec531","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:33:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7a9e585441e54ed9f6b5edf20a37a9c69939a688","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-15 16:34:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c0b0fe0fc11492895277a8a1e1de1bd : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d48de6a4823ef843401b96eeb8d619cc2797ecd8","revisions":{"258eec5e9b9908eaa724816142bea09e1a05df04":{"kind":"REWORK","_number":1,"created":"2022-01-10 05:23:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/823933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/33/823933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/1"}}},"commit":{"parents":[{"commit":"8cbc78dcf47720696fb4b859d92492611a9fa569","subject":"Prepare Yoga M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8cbc78dcf47720696fb4b859d92492611a9fa569"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:22:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:22:40.000000000","tz":540},"subject":"Sync REST_API_REQUIRED_SETTINGS from horizon\u0027s default","message":"Sync REST_API_REQUIRED_SETTINGS from horizon\u0027s default\n\n... to fix some missing items.\n\nChange-Id: Ie3c13d96e8219b4d80b0641f4def49b858f0513a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/258eec5e9b9908eaa724816142bea09e1a05df04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/258eec5e9b9908eaa724816142bea09e1a05df04"}]},"branch":"refs/heads/master"},"45b24103aea893013124f9ec3ac7e382937c9af8":{"kind":"REWORK","_number":2,"created":"2022-01-10 05:31:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/823933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/33/823933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/2"}}},"commit":{"parents":[{"commit":"8cbc78dcf47720696fb4b859d92492611a9fa569","subject":"Prepare Yoga M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8cbc78dcf47720696fb4b859d92492611a9fa569"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:22:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:30:55.000000000","tz":540},"subject":"Do not hard-code REST_API_REQUIRED_SETTINGS","message":"Do not hard-code REST_API_REQUIRED_SETTINGS\n\nThe REST_API_REQUIRED_SETTINGS parameter defines a list of settings\navailable to the client side. Basically this parameter should be\ndefined according to the requirement by Horizon\u0027s implementation.\n\nThis change removes the current hard-code, because the default value\nis now defined in Horizon itself. What is worse, the current value is\noutdated and doesn\u0027t include some items, which would cause problems\nwith behavior.\n\nChange-Id: Ie3c13d96e8219b4d80b0641f4def49b858f0513a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/45b24103aea893013124f9ec3ac7e382937c9af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/45b24103aea893013124f9ec3ac7e382937c9af8"}]},"branch":"refs/heads/master"},"12cb8339154e17f980415ac602eb970b5c2ad883":{"kind":"REWORK","_number":3,"created":"2022-01-10 05:36:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/823933/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/33/823933/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/3"}}},"commit":{"parents":[{"commit":"8cbc78dcf47720696fb4b859d92492611a9fa569","subject":"Prepare Yoga M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8cbc78dcf47720696fb4b859d92492611a9fa569"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:22:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:36:25.000000000","tz":540},"subject":"Do not hard-code REST_API_REQUIRED_SETTINGS","message":"Do not hard-code REST_API_REQUIRED_SETTINGS\n\nThe REST_API_REQUIRED_SETTINGS parameter defines a list of settings\navailable to the client side. Basically this parameter should be\ndefined according to the requirement by Horizon\u0027s implementation.\n\nThis change removes the current hard-code, because the default value\nis now defined in Horizon itself. What is worse, the current value is\noutdated and doesn\u0027t include some items, which would break Horizon\u0027s\nfunctionality.\n\nCloses-Bug: #1956904\nChange-Id: Ie3c13d96e8219b4d80b0641f4def49b858f0513a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/12cb8339154e17f980415ac602eb970b5c2ad883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/12cb8339154e17f980415ac602eb970b5c2ad883"}]},"branch":"refs/heads/master"},"d48de6a4823ef843401b96eeb8d619cc2797ecd8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-10 07:34:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/823933/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/33/823933/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/33/823933/4"}}},"commit":{"parents":[{"commit":"e8882978671686f9a79b8cff58210e393cbee157","subject":"Merge \"Remove usage of custom os_package_type fact\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e8882978671686f9a79b8cff58210e393cbee157"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 05:22:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:34:08.000000000","tz":0},"subject":"Do not hard-code REST_API_REQUIRED_SETTINGS","message":"Do not hard-code REST_API_REQUIRED_SETTINGS\n\nThe REST_API_REQUIRED_SETTINGS parameter defines a list of settings\navailable to the client side. Basically this parameter should be\ndefined according to the requirement by Horizon\u0027s implementation.\n\nThis change removes the current hard-code, because the default value\nis now defined in Horizon itself. What is worse, the current value is\noutdated and doesn\u0027t include some items, which would break Horizon\u0027s\nfunctionality.\n\nCloses-Bug: #1956904\nChange-Id: Ie3c13d96e8219b4d80b0641f4def49b858f0513a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d48de6a4823ef843401b96eeb8d619cc2797ecd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d48de6a4823ef843401b96eeb8d619cc2797ecd8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
