)]}'
{"id":"openstack%2Fpuppet-horizon~824024","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fwallaby~I1f67c00a737ebea6f965c92a923652632641528d","project":"openstack/puppet-horizon","branch":"stable/wallaby","topic":"default_availability_zone-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-12 17:43:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f67c00a737ebea6f965c92a923652632641528d","subject":"Handle string value for LAUNCH_INSTANCE_DEFAULTS","status":"MERGED","created":"2022-01-11 11:35:00.000000000","updated":"2022-01-12 17:43:31.000000000","submitted":"2022-01-12 17:43:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824024-default_availability_zone-stable/xena-stable/wallaby","cherry_pick_of_change":824023,"cherry_pick_of_patch_set":1,"meta_rev_id":"86327adcb05897575f40e4e885155b6eaba0ae84","_number":824024,"virtual_id_number":824024,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-12 17:43:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-01-11 14:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-01-12 15:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-01-12 15:16:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 14:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 14:30:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-01-12 15:16:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c95dac379d55c543ca3e676fdf59cb31537455b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-11 11:35:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"8bac8a4cc014aa52e5705dfedc6bda7559f894f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 14:04:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec38ed4fa704330967f33563e0bc17c : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f03f2fe76c54a32b9254d77d6a9c368 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/83d9a619c7244e739e8910bc8040e7b6 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/730f75b887e44b97aeae663079f78887 : SUCCESS in 12m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45b1c2316b44436c9a63c6272e6de48c : SUCCESS in 11m 00s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85da077131bf4357b62823fb9379ee7b : FAILURE in 1h 02m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffc5b14c7b744677a18a4a20d39b36b0 : SUCCESS in 1h 39m 00s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec4f926457d24f0798b1b5e42bbb8a60 : SUCCESS in 16m 43s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/017a0137dded44218edd12dc641b7d16 : SUCCESS in 11m 32s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c9f7f19504c745f9b8cbe070ec169638 : SUCCESS in 2h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8173e6781d23f477d583aff0c9bae959460d2b18","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-11 14:30:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4f38a0abfccb36cf5aa9bbd5056ed88476e4e19","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-12 15:16:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"257a8a54f596453cb5fcd7d0aebd5679a90b5c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 15:18:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d233c1806b108e3879ea58589722adcc47fa16f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:43:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ca66d4cf7894018b7caceeab4cc3682 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f089ef650ef4b718ee00918b864148e : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/08dc2ade7e954d5a8accf57e4bcfdfba : SUCCESS in 18m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3c466ed8ee642ebaff3631ea9102dba : SUCCESS in 15m 23s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/07eecf85371c4e31b240e83a62bf8974 : SUCCESS in 1h 24m 11s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2d22ff8843d467bace53bf784663417 : SUCCESS in 22m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83977d0a6ccf45afa1f617ce739338c3 : SUCCESS in 13m 07s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4df7399297384e148d0d667d98e65e55 : SUCCESS in 1h 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"86327adcb05897575f40e4e885155b6eaba0ae84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:43:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d93c965e62c7c72fc1da7d1ea99509c45b4a070","revisions":{"0d93c965e62c7c72fc1da7d1ea99509c45b4a070":{"kind":"REWORK","_number":1,"created":"2022-01-11 11:35:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/824024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/24/824024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/24/824024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/24/824024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/24/824024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/24/824024/1"}}},"commit":{"parents":[{"commit":"8437bc4f341e8ec584c32bc50a1d9301b39ed1f1","subject":"Wallaby and older: Use \u0027UNSET\u0027 to remove heat dashboard parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8437bc4f341e8ec584c32bc50a1d9301b39ed1f1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 04:50:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-11 11:35:00.000000000","tz":0},"subject":"Handle string value for LAUNCH_INSTANCE_DEFAULTS","message":"Handle string value for LAUNCH_INSTANCE_DEFAULTS\n\nDuring Wallaby cylce, the new default_availablility_zone parameter\nwas added to LAUNCH_INSTANCE_DEFAULTS[1]. This parameter takes a string\nvalue and the value should be surrounded by \u0027 (or \") when it is put\ninto local_settings.\n\nThis change makes sure that a string value is properly formatted.\n\n[1] fc8603a499fb3302aeb2ebef714420e0fdcbdfef\n\nChange-Id: I1f67c00a737ebea6f965c92a923652632641528d\n(cherry picked from commit f20fa4ab83212d92cdeff415c3ef500c3b7c65e4)\n(cherry picked from commit 8b9d999bf212aa65f4ff23920df85347946be346)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/0d93c965e62c7c72fc1da7d1ea99509c45b4a070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/0d93c965e62c7c72fc1da7d1ea99509c45b4a070"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
