)]}'
{"id":"openstack%2Fpuppet-horizon~830558","triplet_id":"openstack%2Fpuppet-horizon~master~Ieedc84243326e93d06bb94a121a88beac7213268","project":"openstack/puppet-horizon","branch":"master","topic":"django-launch-instance","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-02 12:51:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieedc84243326e93d06bb94a121a88beac7213268","subject":"Remove LAUNCH_INSTANCE_*_ENABLED from local_settings","status":"MERGED","created":"2022-02-23 08:59:16.000000000","updated":"2022-03-02 12:51:58.000000000","submitted":"2022-03-02 12:51:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830558-django-launch-instance","meta_rev_id":"8dda3ae5feae196eac721e6767e36569605c5ec7","_number":830558,"virtual_id_number":830558,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 12:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-01 22:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-02 10:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-02 10:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 10:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 22:39:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-02 10:43:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"a842dde1f6a54943811ad947e4070a1749ec1bc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 08:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3683589d816bd95f7fda62b80ffaf705a67f2676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 10:43:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2611977ba7f848e48df569a0a35bf305 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7832eabab4e48dd8aa7f7cb9d37fe48 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e02fd101ae64d70b7afbd2fbd92e715 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a082ef34ad1a43038dde188f6a715015 : SUCCESS in 18m 45s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c50faaac1bd415090e9222f842f4e3e : SUCCESS in 19m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4d99dfcc7e48f4a7bde54c7ad20c53 : SUCCESS in 15m 44s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e086f1a5a16f4d5ebcdea9b0fa0fa00a : SUCCESS in 15m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d3a99cd55484738b21263d3f7ba5f2b : SUCCESS in 13m 56s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6d7420d4e6c40a6b8137c3f5f2d857a : FAILURE in 35m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/88bdf7ac036345a88ee5591dbe6499e7 : SUCCESS in 1h 15m 26s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f041b5c846643b9a827932c7698b096 : SUCCESS in 1h 12m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67f7e26b5528468880184a5b04d7b6e3 : SUCCESS in 9m 23s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a5c7215853f490f9644f156ab6ad47b : SUCCESS in 16m 59s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fef3ea1d9568450a81a38d726f8ef1d4 : FAILURE in 8m 28s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c0a3ea8074042fa98a4c2f6b0a39e16 : SUCCESS in 10m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3231e12e737a450bad275d241617d501 : SUCCESS in 9m 50s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3403f5f32254523b5cfd114bac1dab9 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"908a33e0f6433a8d6fdc599056cd6c74b59bcda5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-01 22:39:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30b78a8c246f4b5b062744387d8c73ab8331c7ac","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-02 10:43:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93b6c1fc5d22c0c961932d4f31486e4ffc628ed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:45:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c57de0e942ac6e910f3844d094266b408c3c6335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:51:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b83230a3c704f91bbc5d590517339a2 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d04910224a8b465991fa46f36c6f8390 : SUCCESS in 5m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64ea5f723599406594a9e646f4885621 : SUCCESS in 6m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/08937ba4483b4c14a3c53987cb3c1853 : SUCCESS in 19m 18s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/709ff577207b471da290c6eaffec6044 : SUCCESS in 20m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7d06a06edc0495ba9bd8dedc47853ca : SUCCESS in 18m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137b4972dd6345a3a6559ac1af18416b : SUCCESS in 16m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b957a1b3f0d94262975cb90f14ede91c : SUCCESS in 1h 11m 57s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f7f1c226f7f942df92061600ea136aba : SUCCESS in 18m 10s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f3c1f10b22724f78a06ae5c648e19070 : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8dda3ae5feae196eac721e6767e36569605c5ec7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:51:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"156f4662da83d19f5e54fc272d296601b029dae8","revisions":{"156f4662da83d19f5e54fc272d296601b029dae8":{"kind":"REWORK","_number":1,"created":"2022-02-23 08:59:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/830558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/58/830558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/830558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/830558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/58/830558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/58/830558/1"}}},"commit":{"parents":[{"commit":"20d6e15600b3d598ba92d9ca6be64b989291a4c4","subject":"Merge \"TripleO jobs: Replace CentOS 8 jobs by CentOS 9 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/20d6e15600b3d598ba92d9ca6be64b989291a4c4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 08:57:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 08:58:19.000000000","tz":540},"subject":"Remove LAUNCH_INSTANCE_*_ENABLED from local_settings","message":"Remove LAUNCH_INSTANCE_*_ENABLED from local_settings\n\n... because these parameters have been removed from Horizon by [1].\n\n[1] 6ac31e0ba8df520133d20d21e35085d9b882c04e\n\nChange-Id: Ieedc84243326e93d06bb94a121a88beac7213268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/156f4662da83d19f5e54fc272d296601b029dae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/156f4662da83d19f5e54fc272d296601b029dae8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
