)]}'
{"id":"openstack%2Fpuppet-horizon~840552","triplet_id":"openstack%2Fpuppet-horizon~master~I921a913c51801083a2f799ab3d0d1debc62c8f2a","project":"openstack/puppet-horizon","branch":"master","topic":"horizon-system-scope","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-11 10:54:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I921a913c51801083a2f799ab3d0d1debc62c8f2a","subject":"Support the SYSTEM_SCOPE_SERVICES parameter","status":"MERGED","created":"2022-05-05 00:48:06.000000000","updated":"2022-05-11 10:54:04.000000000","submitted":"2022-05-11 10:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840552-horizon-system-scope","meta_rev_id":"9767f91c060dbad79ddf4d366aa350d84b149e0d","_number":840552,"virtual_id_number":840552,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-11 10:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-05-06 20:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-05-11 09:10:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-05-11 09:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 01:57:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 20:54:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-05-11 09:10:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"32ed3ff07c9e5a2e74506a8fcc9dfe342b681370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-05 00:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"914a87001c6aea4b5ecc00a5cce2a24a25281ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 01:57:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b6e91d74df4fd5a99bf314cf013aab : SUCCESS in 2m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7f131060c8d4e4f826194354edcd8b2 : FAILURE in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53281e2a10a448cd82aa7f754f3f00c2 : FAILURE in 2m 37s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/65898d19570e490e9106ba802d969264 : FAILURE in 12m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7180a67f6e5d4dd1b87d9d4480492bb8 : FAILURE in 10m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7276ade2250434cabc9c8046221dfab : FAILURE in 11m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9343148d46b340d1afdae883edb45850 : FAILURE in 13m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3d8b2f166d34ddf8f0fabd63eeeb7e0 : FAILURE in 7m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3d636ea6b214d8d98bdbc4d419bc641 : FAILURE in 9m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d6b1b53ad2e40c598a8cb5ba23d7a0c : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/87c18e0cd3a74c20b1770c5002a87ca8 : FAILURE in 9m 54s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b781913ef01149349baeceab300c922d : FAILURE in 7m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bfd708c80e54b609b43878755e17eda : FAILURE in 7m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b8946f5ef3b47fcbc35f8a6920b5ffc : FAILURE in 6m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e2724143c6d421d8bed1d2658709b4d : SUCCESS in 2m 53s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e3fe17560097496286b57a4e9705d977 : FAILURE in 1h 06m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6aecb3307c256336f23627d39d72a555b03b79f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-05 03:17:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"983ba93011c38e9d5253d8a77ceb125de3a59ec5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-05 03:20:02.000000000","message":"Topic set to system-scope-services","accounts_in_message":[],"_revision_number":2},{"id":"0537b9d3fc5fa272f79f00f817caf72c686d303e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-05 03:20:38.000000000","message":"Topic system-scope-services removed","accounts_in_message":[],"_revision_number":2},{"id":"56ed029102f6d25bc8cb49d6d099531937b237eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-05 03:20:43.000000000","message":"Topic set to horizon-system-scope","accounts_in_message":[],"_revision_number":2},{"id":"56d8aea120070503bbefdc34a0f77f13e2217a87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 04:47:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b412fbf822ae49e887ce79c6e6174b86 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78fa4bdfe9844aceb179ece104902ce3 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16360bba443b47af8f8ad5a080f01fac : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/075d7e8c29884271affc8f335270b301 : SUCCESS in 16m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b0c4d09f9124bffacce372806b698a6 : SUCCESS in 13m 14s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a101f3b062834da29b913ff21a1d1750 : SUCCESS in 13m 01s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e954be6d77b14a01867dd781b695a8c0 : SUCCESS in 13m 11s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70fd012f04df4ad5b6d228374e81769a : FAILURE in 30m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/835bb79228fc49d581ad56be94e1da1f : FAILURE in 10m 48s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae0cbf08d0e1467cb1787f126f895e8e : SUCCESS in 9m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca987c72f6004dab9f38dc97a41a94d0 : SUCCESS in 15m 50s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c3f1f03209e4ab4a70da8e44411faa4 : FAILURE in 9m 18s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d05e6400ed0e4151aeb4dc8067dc0d53 : SUCCESS in 9m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d071cff32fc04f20aae22e44c8ef31a6 : SUCCESS in 10m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d246ad789f24345b63b32f5ae130708 : SUCCESS in 3m 00s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5d6f77a99c32494c957db6eb47ebe498 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5d6de9da8829b8eeb127433c6cf8bdd6e03538b0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-06 02:23:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7de802e5171b1b586a8293d899f2b96211fbf01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 04:08:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/118870aef8bc430cb9185f96cc5c5dd7 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139e76ee9a37442d9ab2012761aa908c : SUCCESS in 6m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0cbce9703654affaf07d44cfa018505 : SUCCESS in 5m 18s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/490797ca7e15433593f4ff7573756d30 : SUCCESS in 19m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/591b0f972eb8410e8e9f34c742602500 : SUCCESS in 16m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/287baaba94b44053b266dc7ac15a53ee : SUCCESS in 16m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1acafe4e2274935be0080a230755207 : SUCCESS in 15m 36s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23bc6fcb32b740268b4174e8a15ea1e5 : FAILURE in 39m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1a719cfcb04419a878b33e42cab4d5b : SUCCESS in 1h 05m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c44d28b7d94e0b8a6d6c8ebaa56a27 : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7d1a19078b244afb3f7d7628b227183 : SUCCESS in 18m 09s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/49068ccec5f94fed839f22844ce09a13 : FAILURE in 9m 04s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334ed4dbcfee465393b2a03268cb4b17 : SUCCESS in 12m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc0e6e2e8e6d480a83149cdf78b2f3e5 : SUCCESS in 12m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f85a1ad3e1654facbf64a23179f66ed7 : SUCCESS in 3m 50s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d7ec885ccf254759ac8ea725bfde9a7f : SUCCESS in 1h 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ffad88ca56102fad7e55c664da499db63e8a100b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-05-06 20:54:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8fe0e214636af536676a03395c38cdb3bfab67b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-11 09:10:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b1a98fa247310df0c5f7073c52e0cfc8db2baa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 09:11:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"011d2233f32375dcb6d710189e86b5051be621a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:54:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c897a8be21c443bb225180b4288cd1c : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9661573b748c4af1a01c0e85df47da4b : SUCCESS in 3m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e3d8ec086c54abd946db7fb0dfb2023 : SUCCESS in 5m 00s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/465f99b04e574b4e9955e4a5af4d2536 : SUCCESS in 20m 00s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a70053fc42b74c398b08762096d083af : SUCCESS in 11m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6713654c7a464726888ccf55bd4a3ea2 : SUCCESS in 17m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf16692c9ed643e281e59ca14921c7ab : SUCCESS in 51m 09s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/bdabe5371e2149b1a695f2cf4312f37b : SUCCESS in 18m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98adfc0d94534183a2a091ccb70c55ef : SUCCESS in 3m 19s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0eb07588e7914773b020f5fa34fb7a75 : SUCCESS in 1h 10m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9767f91c060dbad79ddf4d366aa350d84b149e0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 10:54:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f6b4b33a4d309522cd06f3e938acf66f34082e0","revisions":{"954827d2072e69ade3b630cdd2a8cddd4356006c":{"kind":"REWORK","_number":1,"created":"2022-05-05 00:48:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/840552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/52/840552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/1"}}},"commit":{"parents":[{"commit":"00ce4ba698b6c8c2c10628558726462cdd6bb63d","subject":"Merge \"Allow customizing apache::vhost parameters for SSL vhost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/00ce4ba698b6c8c2c10628558726462cdd6bb63d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-05 00:46:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-05 00:46:53.000000000","tz":540},"subject":"Support the SYSTEM_SCOPE_SERVICES parameter","message":"Support the SYSTEM_SCOPE_SERVICES parameter\n\nThe SYSTEM_SCOPE_SERVICES parameter was introduced during stable/yoga,\nto enable usage of system scope tokens per service. This change adds\nsupport for the parameter so that users can customize the parameter\nusing this module.\n\nChange-Id: I921a913c51801083a2f799ab3d0d1debc62c8f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/954827d2072e69ade3b630cdd2a8cddd4356006c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/954827d2072e69ade3b630cdd2a8cddd4356006c"}]},"branch":"refs/heads/master"},"3f6b4b33a4d309522cd06f3e938acf66f34082e0":{"kind":"REWORK","_number":2,"created":"2022-05-05 03:17:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/840552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/52/840552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/52/840552/2"}}},"commit":{"parents":[{"commit":"00ce4ba698b6c8c2c10628558726462cdd6bb63d","subject":"Merge \"Allow customizing apache::vhost parameters for SSL vhost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/00ce4ba698b6c8c2c10628558726462cdd6bb63d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-05 00:46:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-05-05 03:17:00.000000000","tz":0},"subject":"Support the SYSTEM_SCOPE_SERVICES parameter","message":"Support the SYSTEM_SCOPE_SERVICES parameter\n\nThe SYSTEM_SCOPE_SERVICES parameter was introduced during stable/yoga,\nto enable usage of system scope tokens per service. This change adds\nsupport for the parameter so that users can customize the parameter\nusing this module.\n\nChange-Id: I921a913c51801083a2f799ab3d0d1debc62c8f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/3f6b4b33a4d309522cd06f3e938acf66f34082e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/3f6b4b33a4d309522cd06f3e938acf66f34082e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
