)]}'
{"id":"openstack%2Fpuppet-horizon~853334","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fwallaby~Iee260f8578ed6c473c651e0665a548c68e94c954","project":"openstack/puppet-horizon","branch":"stable/wallaby","topic":"bug/1983182","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-18 17:34:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee260f8578ed6c473c651e0665a548c68e94c954","subject":"Add inet6 prefix only when MemcacheCache is used","status":"MERGED","created":"2022-08-17 05:43:32.000000000","updated":"2022-08-18 17:34:46.000000000","submitted":"2022-08-18 17:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853334-bug/1983182","cherry_pick_of_change":853213,"cherry_pick_of_patch_set":1,"meta_rev_id":"f0809df8331813648f49b9abd518bf3122888ea9","_number":853334,"virtual_id_number":853334,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 17:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-17 20:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-18 15:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-18 15:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 08:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 20:52:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-18 15:16:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"74515d9f6bb82fd51fedf825bd705c33f1728ed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 05:43:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"1f39ae9c8859ba051f234e9a3215f944846d30bb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 05:43:43.000000000","message":"Topic bug/1983182-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"6f737c9142243f27bb128b264217aca64a78e7b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-17 05:43:46.000000000","message":"Topic set to bug/1983182","accounts_in_message":[],"_revision_number":1},{"id":"13d7843fec74d540520423ccfb72772eaada8c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 08:12:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2a7ba5c8d343d383fceec30f14407c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51780937b8ec475ea2cf78f9f4132e2b : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/281a4c9b56b44385bbbaac125cf39454 : SUCCESS in 3m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a80bb802acad48fb8e2e9637d2c23e49 : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79b073100ac34e339f72bb0bb799e05b : SUCCESS in 12m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0947cf5cfd345de92ba97fcc3cfa944 : SUCCESS in 11m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d2a2e78ae5948b58888aa9015456bf7 : FAILURE in 38m 30s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b5a271044184f67bf29787b8ac93012 : SUCCESS in 1h 43m 26s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ded321e55684287bc09366daaf521d5 : SUCCESS in 18m 31s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8534703bc56e43e984f5038cc3ece434 : SUCCESS in 9m 30s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/706991f7289e4488970c4415cb3f8624 : SUCCESS in 2h 21m 36s\n- puppet-horizon-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fcd3d7497cdf45298a4ad5d79edbdd32 : SUCCESS in 1h 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b701f92674d097a498382d5260c6a0d5f18d53e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-17 20:52:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14b4b6d50ffc15d4e0dc0df500ad2997ffd5f950","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-18 15:16:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc2cd17f66dbd836743d764adf4a916eb1aa18de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 15:18:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78ae2dd21fc5fdc67d8b5db9fc36b0651d887f45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:34:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6a07ae90ea480391789fb52053e171\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d15a46eb4c54a49a01ac09f9eda6274 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89a1c24a80f34634aac3d3edd316b401 : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b89e7e1e2e8d4fd18536ff03997663c7 : SUCCESS in 26m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8355e1fc1534bfbbe2c3d4b822fa585 : SUCCESS in 23m 32s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27d5533a499a4df699756ec136a513cc : SUCCESS in 1h 34m 46s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab880a2186ab4f128438b69bc090e3ce : SUCCESS in 21m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f1e38a3f7ff4d8686415651a34d3572 : SUCCESS in 12m 55s (non-voting)\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c35b0365ce164580ba541d302051bf48 : SUCCESS in 2h 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f0809df8331813648f49b9abd518bf3122888ea9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:34:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e7e2e27be7041c654cf3f9e9785d3e5243128c1","revisions":{"7e7e2e27be7041c654cf3f9e9785d3e5243128c1":{"kind":"REWORK","_number":1,"created":"2022-08-17 05:43:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/853334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/34/853334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/853334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/853334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/853334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/34/853334/1"}}},"commit":{"parents":[{"commit":"dc84cdb9f62d7d2a34cf4ce6252f3c21da945d62","subject":"Accept array for cache_server_url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dc84cdb9f62d7d2a34cf4ce6252f3c21da945d62"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-30 15:22:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-17 05:43:32.000000000","tz":0},"subject":"Add inet6 prefix only when MemcacheCache is used","message":"Add inet6 prefix only when MemcacheCache is used\n\ninet6 prefix is required only by python-memcached and it should NOT be\nadded when a different backend like PyMemcacheCache is used.\n\nConflicts:\n\tspec/classes/horizon_init_spec.rb\n\nCloses-Bug: #1983182\nChange-Id: Iee260f8578ed6c473c651e0665a548c68e94c954\n(cherry picked from commit 94422e6dccfed2d34100f7bd2c7bd87cdc5cfd18)\n(cherry picked from commit bb7e2d373b3837857a81a925087427d5f084d742)\n(cherry picked from commit 54b7eb0579519e006d432b5b96e91cef5fcc9b16)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7e7e2e27be7041c654cf3f9e9785d3e5243128c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7e7e2e27be7041c654cf3f9e9785d3e5243128c1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
