)]}'
{"id":"openstack%2Fpuppet-horizon~855298","triplet_id":"openstack%2Fpuppet-horizon~master~I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd","project":"openstack/puppet-horizon","branch":"master","topic":"bug/1988205","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-01 11:23:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd","subject":"Ensure pymemcache library is installed","status":"MERGED","created":"2022-08-30 23:47:18.000000000","updated":"2022-09-01 11:24:34.000000000","submitted":"2022-09-01 11:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855298-bug/1988205","meta_rev_id":"cea8f58f0ba5eb36500791663745031ea3423adf","_number":855298,"virtual_id_number":855298,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-01 11:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-08-31 14:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-09-01 09:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-09-01 09:13:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 01:02:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 14:38:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-09-01 09:13:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"6f96c387fb7288f013f5ff2ee03bd06a0ed986c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-30 23:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c510f8f5cf6e6c7982d286e882c0979d0f7b6f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-30 23:48:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c789ef82a23b1627c5fae97c3676242b1c27218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 01:02:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0d20d58cfef405b86a9c6d7cf58a20f\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/403097a5d6c74df8857d06677a6392db : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30eac4d944ae48f5aa74b20c66946217 : FAILURE in 3m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4522bc9d7b3f47a280be81be5c8ca50c : FAILURE in 3m 50s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca1f42f8fcc544e7af56f18a4b366ff7 : FAILURE in 18m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ce75d4bf37b4a13b8972e376179aed4 : FAILURE in 13m 14s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b61016c99994a1094dc13f843f16bab : FAILURE in 12m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dac64dd76bd14806911c6a8fa3057a70 : FAILURE in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a903e1c753a8456c8bfba96f4cbe61d0 : FAILURE in 10m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df9f86efcbfd43d2b0ad423f8a8da551 : FAILURE in 17m 34s\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba5e2bf556644827a61fec5c34e9b261 : FAILURE in 10m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88be670383dc44f28aad2918d8f649e9 : FAILURE in 17m 38s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd655fde0cd747b3985a924a6064ea28 : FAILURE in 10m 25s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0faa53c724a04159ab9defbe98adef46 : FAILURE in 16m 38s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4f141e9fe9e4bdf840785b200472fd8 : FAILURE in 10m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1526cb38dac44360a19c172ee3e204ed : FAILURE in 10m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42c8a9adcf864eb5a6ce59703c1ac9e4 : SUCCESS in 4m 04s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4e19f3a2f5fd441a805cffeec8550463 : FAILURE in 1h 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5887322a63dea76aa3d3101ca590b247b7e15321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 01:10:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1eaa2243a6ca2cc6034514fecec5f1a27043d75f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 01:30:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"44a1107ffa17e0f4e60ecabc13d0d10c69691028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-31 01:53:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"efced357747e1793e0b95d2f9d25a40def0b564b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 03:25:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b6f07473df840c49e9dfd3d28ed0552\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65633469ea9e4dd8b14084fd4ed1845b : SUCCESS in 2m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98ecd39586ff42578a1447a70ee69ec9 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7590b22eab9f4b178cddb719d893cd1f : SUCCESS in 3m 15s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/20ebfa7a8e454e62953699786c064450 : SUCCESS in 16m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fb87932245349f8873bccfdbf6a95ad : SUCCESS in 17m 05s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dff1f21a417e490ba6cc2631db73af99 : SUCCESS in 12m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/357a2779c67349ec966c5804dd156af6 : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/239208e472534e81a78b6faa4e3affa7 : FAILURE in 38m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6af6f049bec04246a2f46182eba3c3ba : SUCCESS in 1h 03m 55s\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d4d83b4dde74556b7b699f06bb1a274 : FAILURE in 29m 18s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e71b5bb844bc43fba1dfd7bb7e130616 : SUCCESS in 52m 35s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f84c801891f2469081ac671dd7dc5c55 : SUCCESS in 11m 40s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc00008c44ff4c45a93ba253b8862328 : SUCCESS in 16m 55s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893d0510cac44cd4b605d92e4e26309d : SUCCESS in 12m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de254897a882460abef6b8eb2c70eb6d : SUCCESS in 12m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38052e2bcacd4ee7b29830f03057c58f : SUCCESS in 3m 26s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4326aa6aa2eb453489cfc6449c621c9b : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":5},{"id":"f6d05c8218f0a31ccc20d278ae4626f2430a9573","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-31 14:38:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f52ce27c33ada2a744b0f14eaaaa2d1b027b71f6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-01 09:13:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8d85532988d3e9bfa6148a88a14b8cf106dc7295","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 09:14:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d04de0f79453617c4019eced5887f8ce565bd4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 11:23:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/717976f6157a47a48f301d1a38bdc27c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d3fb8ceb23488bb1a6ac6ac4e32b25 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82aeca8654024e9fba3f7a43aad5ca32 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7f7b53efd4f456ea3985474e55d2edf : SUCCESS in 3m 27s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4eff23504bf54320bf96045cf745ebf9 : SUCCESS in 18m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96fb30643aa4311a18321ec003206f0 : SUCCESS in 22m 36s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/431703ff3b2244e087c6cb47b4b8a211 : SUCCESS in 15m 33s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b33e15cd59146559265256cd95877af : SUCCESS in 1h 12m 38s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db5927139bcd4760a39041d83805180e : SUCCESS in 53m 19s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b92edda44d54282984aa03e3a8d37f3 : SUCCESS in 21m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/759a4c18b65d4258bca68752797b2c51 : SUCCESS in 4m 00s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2615f3ec58040c983706d203e806e97 : SUCCESS in 1h 44m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d1155d0ca1f8164069233eb4e8be4da80fc849e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 11:23:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cea8f58f0ba5eb36500791663745031ea3423adf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 11:24:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba4022c12b1e4a8a9ff4b61e78b1b77d\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/804ef1911aa249adae54ecfdfeffbbc6 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5660159a6e2237c8bcdc66167873b6534da1d935","revisions":{"aa51b5c31a3e98d7b5a9f4862dbd9a0e16c74951":{"kind":"REWORK","_number":1,"created":"2022-08-30 23:47:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/855298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/98/855298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/1"}}},"commit":{"parents":[{"commit":"6a36a3e0530658249d659fc361840eac1ea78d24","subject":"Merge \"Cap upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6a36a3e0530658249d659fc361840eac1ea78d24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:45:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:46:30.000000000","tz":540},"subject":"Ensure pymemcache library is installed","message":"Ensure pymemcache library is installed\n\nThe pymemcache library should be installed when pymemcache backend is\nused but this library is not installed by the package dependencies and\nshould be installed additionally.\n\nThis also ensures the cache libraries are installed during package\ninstallation phase, so that the libraries are available when horizon\nis started.\n\nCloses-Bug: #1988205\nChange-Id: I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/aa51b5c31a3e98d7b5a9f4862dbd9a0e16c74951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/aa51b5c31a3e98d7b5a9f4862dbd9a0e16c74951"}]},"branch":"refs/heads/master"},"223969f9d6efb5efbdfe1b7eeac5ae426fc90f41":{"kind":"REWORK","_number":2,"created":"2022-08-30 23:48:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/855298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/98/855298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/2"}}},"commit":{"parents":[{"commit":"6a36a3e0530658249d659fc361840eac1ea78d24","subject":"Merge \"Cap upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6a36a3e0530658249d659fc361840eac1ea78d24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:45:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:48:25.000000000","tz":540},"subject":"Ensure pymemcache library is installed","message":"Ensure pymemcache library is installed\n\nThe pymemcache library should be installed when pymemcache backend is\nused but this library is not installed by the package dependencies and\nshould be installed additionally.\n\nThis also ensures the cache libraries are installed during package\ninstallation phase, so that the libraries are available when horizon\nis started.\n\nCloses-Bug: #1988205\nChange-Id: I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/223969f9d6efb5efbdfe1b7eeac5ae426fc90f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/223969f9d6efb5efbdfe1b7eeac5ae426fc90f41"}]},"branch":"refs/heads/master"},"c6e2d0696f205025f22bf948fc9b2153253e5261":{"kind":"REWORK","_number":3,"created":"2022-08-31 01:10:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/855298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/98/855298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/3"}}},"commit":{"parents":[{"commit":"6a36a3e0530658249d659fc361840eac1ea78d24","subject":"Merge \"Cap upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6a36a3e0530658249d659fc361840eac1ea78d24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:45:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 01:10:09.000000000","tz":540},"subject":"Ensure pymemcache library is installed","message":"Ensure pymemcache library is installed\n\nThe pymemcache library should be installed when pymemcache backend is\nused but this library is not installed by the package dependencies and\nshould be installed additionally.\n\nThis also ensures the cache libraries are installed during package\ninstallation phase, so that the libraries are available when horizon\nis started.\n\nCloses-Bug: #1988205\nChange-Id: I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c6e2d0696f205025f22bf948fc9b2153253e5261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c6e2d0696f205025f22bf948fc9b2153253e5261"}]},"branch":"refs/heads/master"},"19661cceaaa030fad8744d09ebe368734fdf98b6":{"kind":"REWORK","_number":4,"created":"2022-08-31 01:30:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/855298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/98/855298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/4"}}},"commit":{"parents":[{"commit":"6a36a3e0530658249d659fc361840eac1ea78d24","subject":"Merge \"Cap upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6a36a3e0530658249d659fc361840eac1ea78d24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:45:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 01:29:44.000000000","tz":540},"subject":"Ensure pymemcache library is installed","message":"Ensure pymemcache library is installed\n\nThe pymemcache library should be installed when pymemcache backend is\nused but this library is not installed by the package dependencies and\nshould be installed additionally.\n\nThis also ensures the cache libraries are installed during package\ninstallation phase, so that the libraries are available when horizon\nis started.\n\nCloses-Bug: #1988205\nChange-Id: I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/19661cceaaa030fad8744d09ebe368734fdf98b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/19661cceaaa030fad8744d09ebe368734fdf98b6"}]},"branch":"refs/heads/master"},"5660159a6e2237c8bcdc66167873b6534da1d935":{"kind":"REWORK","_number":5,"created":"2022-08-31 01:53:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/855298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/98/855298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/98/855298/5"}}},"commit":{"parents":[{"commit":"6a36a3e0530658249d659fc361840eac1ea78d24","subject":"Merge \"Cap upper version of puppetlabs-apache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6a36a3e0530658249d659fc361840eac1ea78d24"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-30 23:45:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-31 01:53:13.000000000","tz":540},"subject":"Ensure pymemcache library is installed","message":"Ensure pymemcache library is installed\n\nThe pymemcache library should be installed when pymemcache backend is\nused but this library is not installed by the package dependencies and\nshould be installed additionally.\n\nThis also ensures the cache libraries are installed during package\ninstallation phase, so that the libraries are available when horizon\nis started.\n\nCloses-Bug: #1988205\nChange-Id: I90cd9d9e1276fae71fcb9cbc7ba95b8839690dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5660159a6e2237c8bcdc66167873b6534da1d935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5660159a6e2237c8bcdc66167873b6534da1d935"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
