)]}'
{"id":"openstack%2Fpuppet-horizon~856193","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fyoga~Ie36d70b7b9fdde5f1ceb4be59cfb4f5c16981e7d","project":"openstack/puppet-horizon","branch":"stable/yoga","topic":"puppet-horizon-dashboards-stable/yoga","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-09-09 10:37:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie36d70b7b9fdde5f1ceb4be59cfb4f5c16981e7d","subject":"Add support for independent designate-dashboard configurations","status":"MERGED","created":"2022-09-07 00:56:06.000000000","updated":"2022-09-09 10:39:03.000000000","submitted":"2022-09-09 10:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856193-puppet-horizon-dashboards-stable/yoga","cherry_pick_of_change":830130,"cherry_pick_of_patch_set":5,"meta_rev_id":"4d48f3905ae4d12102d99a3fb497a600bba543a2","_number":856193,"virtual_id_number":856193,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 10:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-09-07 17:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-09-09 09:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-09-09 09:16:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 02:21:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 17:43:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-09-09 09:16:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"29f7d90d5dd6d277dd74118688faefb7f36e2a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 00:56:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b14c6627c540d0281055e31385fbc2f58d3207a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 00:56:17.000000000","message":"Topic puppet-horizon-dashboards-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"8c144dedae3432aea7f71d723dbcf91b67ee2d74","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 00:56:20.000000000","message":"Topic set to puppet-horizon-dashboards","accounts_in_message":[],"_revision_number":1},{"id":"74f3ed1544da29f720e2d85b55616404e1c9784c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 00:56:22.000000000","message":"Topic puppet-horizon-dashboards removed","accounts_in_message":[],"_revision_number":1},{"id":"83746649dffec6bee7ebaa70e5e178d0fd97fbd4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-09-07 00:56:29.000000000","message":"Topic set to puppet-horizon-dashboards-stable/yoga","accounts_in_message":[],"_revision_number":1},{"id":"4a5b670d843aa348ea169f20c58eb6c848cb844a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 02:21:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65bb99d2b9d94378be3ae38dcfbad196\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0f3098fee2f446f93b6b1e75e8b9982 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10140a97362b4c50bd378f93c2b055c9 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5b180ac456740369244ce6407aaa6e8 : SUCCESS in 3m 20s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd068a22cc694ab3b18a87dfe3b1df5b : SUCCESS in 19m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e5ac72c69094270ade6d1d192ef91b0 : SUCCESS in 16m 17s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea932bc48edb48828f7ba1022f754856 : SUCCESS in 17m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d08758148e49e4a11e85a3ab349668 : SUCCESS in 12m 33s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b94adba20ead4277bf7a4508ed850ddb : FAILURE in 46m 34s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ba9a37716de4de4974b368c073d31d2 : SUCCESS in 1h 14m 56s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27db5f0118324b3a8413281a0fd91e2d : SUCCESS in 1h 00m 03s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cccb64807e14b0a98f2d33513001875 : SUCCESS in 12m 39s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec6f3d12783344dd9829907c2ac9ae90 : SUCCESS in 18m 07s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2a10a04aa0543a8b7c8b36201c5d480 : SUCCESS in 19m 55s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e51f3d5d428a4445bf246205a254dca8 : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5adc0ad764be4c328429c48279ce6dee : SUCCESS in 12m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42b90616ef0b41afab39e5d9e449816d : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"eb9b0589aa3318a03ed04f67caf828c97cd956c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-09-07 17:43:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40857de6b690f4ed8f9d5a7c9436c920796cabfa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-09 09:16:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ef8ab58f57bc9e5d73cf7764d3b2e1b65d07bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 09:25:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85868a3a06902fae8d28835f9ad37b65cc356b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 10:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7151a77c4f4be0a8334ffe96f714ea\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24978eb0cea742ddb512065508d945c1 : SUCCESS in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44a70c2811624f6bb1370de54dbb9dae : SUCCESS in 3m 12s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c923ae9a6ebb45f392377ccc1529c40e : SUCCESS in 4m 26s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a846ca4931b47b1952b55c2a847794f : SUCCESS in 22m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69696a60da5f44dc8060178844b8b18b : SUCCESS in 18m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bfbbdde2852442ba6bfeda98770bba4 : SUCCESS in 17m 47s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/55611a2461a6464883c376fbc8eb4ff3 : SUCCESS in 1h 09m 26s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1ebd5781447486091fcf13a7e46d482 : SUCCESS in 22m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6390d307bc954052814150cce79f63e2 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"b75145d3d66900722a6732445ff1c047cb80f4fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 10:37:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d48f3905ae4d12102d99a3fb497a600bba543a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 10:39:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494cc695f98242b5814a5a6d2354ac9a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf2bebd488ee4016a102bcb69bbc7db8 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be7c2640187c4150dbf69ab9c2ea138f86bede3a","revisions":{"be7c2640187c4150dbf69ab9c2ea138f86bede3a":{"kind":"REWORK","_number":1,"created":"2022-09-07 00:56:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/856193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/93/856193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/93/856193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/93/856193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/93/856193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/93/856193/1"}}},"commit":{"parents":[{"commit":"cf626d236763d9c739a41c06b6b891a0b6f6603f","subject":"Do not show diff of local_settings(.py)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/cf626d236763d9c739a41c06b6b891a0b6f6603f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-21 10:05:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-09-07 00:56:06.000000000","tz":0},"subject":"Add support for independent designate-dashboard configurations","message":"Add support for independent designate-dashboard configurations\n\nThis change introduces the new horizon::dashboards::designate class,\nwhich allows users to manage the independent configuration file for\ndesignate dashboard (_1799_dns_settings.py).\n\nDepends-on: https://review.opendev.org/823238\nChange-Id: Ie36d70b7b9fdde5f1ceb4be59cfb4f5c16981e7d\n(cherry picked from commit 7577a23c65492e7d3d0c467609d182af5bce3ea1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/be7c2640187c4150dbf69ab9c2ea138f86bede3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/be7c2640187c4150dbf69ab9c2ea138f86bede3a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
