)]}'
{"id":"openstack%2Fpuppet-horizon~862007","triplet_id":"openstack%2Fpuppet-horizon~master~I99734d23c545a1b03529aa7d99091f540256b157","project":"openstack/puppet-horizon","branch":"master","topic":"reno-zed","hashtags":[],"change_id":"I99734d23c545a1b03529aa7d99091f540256b157","subject":"Update master for stable/zed","status":"MERGED","created":"2022-10-20 11:05:16.000000000","updated":"2022-10-20 14:29:57.000000000","submitted":"2022-10-20 14:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862007-reno-zed","meta_rev_id":"16b63b756fd6810e8030a4d88de42bde1d4157d7","_number":862007,"virtual_id_number":862007,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 14:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-10-20 14:19:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-10-20 14:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:18:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 14:19:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"2e48bd7c416c1ec63066750a9118e68b661846de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2022-10-20 11:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d2f9af5258e53d6900c2d40f6c2a65e2530be37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/104b599561944774b8c113088673531a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/649b712737d24f7793dda5a286ff6667 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"553de3b8ed2aceaacf3a83344d2a9912205a4b3e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-20 14:19:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db0b4d7c0eeb31e09cce223c32a60f703cbb614f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:20:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d09c1e107863825b792574d8210067faca42806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:28:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2933831fbbff46cda9d372f6a00a1bee\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/caf61b2b73e348baabc3429751fe5926 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5c2e98c2f7d419b6ad959b33fa0d2aa1f8c189a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:28:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16b63b756fd6810e8030a4d88de42bde1d4157d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 14:29:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e48b32ad7aab4e649fa7e323284c97bc\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89f3e8abc7154e7fb1ea55f2e2f2e0c4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5fd390cb80fba789add203d7979854e1163bb8b","revisions":{"e5fd390cb80fba789add203d7979854e1163bb8b":{"kind":"REWORK","_number":1,"created":"2022-10-20 11:05:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/07/862007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/07/862007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/07/862007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/07/862007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/07/862007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/07/862007/1"}}},"commit":{"parents":[{"commit":"dac3cece5198e9cd76c7419dc697a34d85b62d08","subject":"Prepare Zed RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dac3cece5198e9cd76c7419dc697a34d85b62d08"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-10-20 11:05:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2022-10-20 11:05:14.000000000","tz":0},"subject":"Update master for stable/zed","message":"Update master for stable/zed\n\nAdd file to the reno documentation build to show release notes for\nstable/zed.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/zed.\n\nSem-Ver: feature\nChange-Id: I99734d23c545a1b03529aa7d99091f540256b157\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e5fd390cb80fba789add203d7979854e1163bb8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e5fd390cb80fba789add203d7979854e1163bb8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
