)]}'
{"id":"openstack%2Fpuppet-horizon~863350","triplet_id":"openstack%2Fpuppet-horizon~stable%2Fwallaby~Iea08dac23cab190bed38cfafb230160648bb94a3","project":"openstack/puppet-horizon","branch":"stable/wallaby","topic":"lp/1966382-stable/wallaby","attention_set":{},"removed_from_attention_set":{"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-11-04 01:30:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea08dac23cab190bed38cfafb230160648bb94a3","subject":"setuptools: Disable auto discovery","status":"MERGED","created":"2022-11-02 15:14:51.000000000","updated":"2022-11-04 01:30:26.000000000","submitted":"2022-11-04 01:30:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863350-lp/1966382-stable/wallaby","cherry_pick_of_change":863225,"cherry_pick_of_patch_set":2,"meta_rev_id":"66a7e9bc9510b63b821725dbfc1ad014627c1134","_number":863350,"virtual_id_number":863350,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-04 01:30:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-04 01:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-04 01:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 15:14:51.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"CC"},{"updated":"2022-11-02 15:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-02 16:21:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-03 05:42:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2022-11-03 05:46:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REMOVED"}],"messages":[{"id":"8b10b20208e02d7cba14f58a8b9a0f1a53f5f4da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-02 15:14:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* setup.py","accounts_in_message":[],"_revision_number":1},{"id":"e24f832fe3fd5826cd6fb7ccbeadd5a87ce4c428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 15:26:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df2e60e2cf94df295ae14b2531c86e0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c066fbe9ce5243d28f6934d22058c125 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7a03db09b833026ff69a7927122db6a62b717727","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-02 16:21:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f31790fd68f0b1ae219a1f88b323a106ec961684","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-02 16:25:50.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"7042c89d7fb50c93a0a0fede9d69aafe1c718186","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-03 05:42:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"393817189f7eff93058f4b4756c1b62d40cb082a","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-03 05:46:47.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_32926\u003e.","accounts_in_message":[{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"_revision_number":2},{"id":"5c9d4760398b35d01693a19086053b076fbc5587","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-03 05:46:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"44c7146a5409a9b4b8de6c9070ae1b607d7b283a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 05:49:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/381b40a732c14ab7bc778c2e5c0b7aac\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/485adc63a86c40a29ee219fcb08d9583 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c8a5c5c99d2efaa4101cc4852a67a810bcdcaaf4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-04 01:26:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f0d27322b8d0c261284b36ac7fdf39d79b98297b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:26:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbd958fb217d646a8ee2355fb478076fe3022069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:30:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98f233135a954a28b5f04dbb12cacce6\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7307acfe0db4219adfdec3ab55a7b4e : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":2},{"id":"66a7e9bc9510b63b821725dbfc1ad014627c1134","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 01:30:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e768b149d5f6f3aa1627134fba7a0b9e9667ad1f","revisions":{"f4048641416c3937c11980a2c8bac85b71436713":{"kind":"REWORK","_number":1,"created":"2022-11-02 15:14:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/863350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/50/863350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/1"}}},"commit":{"parents":[{"commit":"2f3476b43190ebea06efdf1e3768123563d45493","subject":"Prepare the final wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2f3476b43190ebea06efdf1e3768123563d45493"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:38:50.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-02 15:14:51.000000000","tz":0},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iea08dac23cab190bed38cfafb230160648bb94a3\n(cherry picked from commit 71e676706a6619572ac09c8f9eea67b6715ef9a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f4048641416c3937c11980a2c8bac85b71436713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f4048641416c3937c11980a2c8bac85b71436713"}]},"branch":"refs/heads/stable/wallaby"},"e768b149d5f6f3aa1627134fba7a0b9e9667ad1f":{"kind":"REWORK","_number":2,"created":"2022-11-03 05:42:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/863350/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/50/863350/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/50/863350/2"}}},"commit":{"parents":[{"commit":"2f3476b43190ebea06efdf1e3768123563d45493","subject":"Prepare the final wallaby release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2f3476b43190ebea06efdf1e3768123563d45493"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-03-29 08:38:50.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-03 05:41:46.000000000","tz":540},"subject":"setuptools: Disable auto discovery","message":"setuptools: Disable auto discovery\n\nThe latest release of setuptools 61.0 made a breaking change[1] and\nbecause of this change \u0027pip install\u0027 fails with the following error.\n\n~~~\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027lib\u0027, \u0027spec\u0027, \u0027manifests\u0027, \u0027releasenotes\u0027].\n~~~\n\nUsers that don\u0027t set \u0027packages\u0027, \u0027py_modules\u0027, or configuration\u0027\nare still likely to observe the auto-discovery behavior, which may\nhalt the build if the project contains multiple directories and/or\nmultiple Python files directly under the project root.\n\nTo disable auto discovery, one can do below in setup.py\n\n~~~\nsetuptools.setup(..,packages\u003d[],..)\n~~~\n\nor\n\n~~~\nsetuptools.setup(..,py_modules\u003d[],..)\n~~~\n\n[1] https://github.com/pypa/setuptools/issues/3197\n\nNote setup.py is not used to install puppet modules. However it is used\nto generate a release note, thus should be fixed.\n\nConflicts:\n\tsetup.py\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iea08dac23cab190bed38cfafb230160648bb94a3\n(cherry picked from commit 71e676706a6619572ac09c8f9eea67b6715ef9a5)\n(cherry picked from commit 6baf1f30519531ca8778fe2ce15894fcb2f75e11)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e768b149d5f6f3aa1627134fba7a0b9e9667ad1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e768b149d5f6f3aa1627134fba7a0b9e9667ad1f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
