)]}'
{"id":"openstack%2Fpuppet-horizon~874048","triplet_id":"openstack%2Fpuppet-horizon~master~I376af3b7c4525036bf1bb449cb71490495395f2c","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-20 01:38:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I376af3b7c4525036bf1bb449cb71490495395f2c","subject":"Deprecate ineffective images_panel","status":"MERGED","created":"2023-02-16 03:08:45.000000000","updated":"2023-02-20 05:41:22.000000000","submitted":"2023-02-20 05:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874048","meta_rev_id":"2a585fd099dc48e56cc279b40a81586d5da32173","_number":874048,"virtual_id_number":874048,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-20 05:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-02-16 14:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-17 11:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-17 11:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 05:03:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 14:48:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-17 11:19:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"e5fe4f7f7b8da248614323e4004576d8e7ed6eab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-16 03:08:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef76f25d5c44a351b7759d085e464fd90da47a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 05:03:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7b57ff8d27646329470b838646e5ffd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/465355deac3f43d9bee9f9277450b303 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b57392e1849643aea0ad3a2b5f2f03d5 : SUCCESS in 3m 24s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b01fce8f15641b48125d8d2422ff59e : POST_FAILURE in 17m 22s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7362e3aaefc47d4b3b51689d484b7aa : SUCCESS in 12m 34s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97bf7af0ea8c4ca39a98d1cdc14e4549 : SUCCESS in 11m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33cfc329f4d24466b10dd56b0106e7ab : FAILURE in 30m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abe61e3dfa074644a68c6904e156a1cc : SUCCESS in 59m 19s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc78cc93d55402b886cd04c482a85fd : SUCCESS in 38m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cdf0923e95f495ba6542bf3b76eb4e9 : SUCCESS in 51m 31s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/de788c173aa645ff9d1252a950eef2a4 : SUCCESS in 19m 35s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90daf7ef012347b998ccda155db7bcb9 : SUCCESS in 9m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c77572acfab84a58b23e0dd893a518b5 : SUCCESS in 10m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c98e2028844767ab409e4e30822970 : SUCCESS in 4m 05s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cc6995cf072f41599e23c99d38a22bab : SUCCESS in 1h 42m 57s","accounts_in_message":[],"_revision_number":1},{"id":"982614f6496d2f53a6702252448a9f4a6c224b61","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-16 07:44:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27464c847200ce84a2f88833c830ac2fda3b77a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 09:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc484f1baa7f4d3087522683006c0f02\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a975ef2837548c5b84170163b6d7b35 : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25f5ca0587e245c5a66bf7f16bd8c596 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5a2db7dfb084e8aa817e7494e41051e : SUCCESS in 13m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56b8718700be434abd45681ad3c3356d : SUCCESS in 10m 45s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96d586d15f39429bae4a068001b71182 : SUCCESS in 11m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107df50d3bf64ca89b9b70a9e8c363b8 : FAILURE in 26m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f6e321062974c7d86680b6da1c7e94c : SUCCESS in 1h 00m 37s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99c03934659c4f529c36c6b1bbfc78ae : SUCCESS in 43m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ddbedc315734a6986925694e0094cb2 : SUCCESS in 56m 37s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d432cf105b04cadb3f9d77390834587 : SUCCESS in 13m 11s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac880d8cc8784dce99f0132ad653cb01 : SUCCESS in 8m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44a3274276954092a02ac46eec77d70d : SUCCESS in 9m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d81619ca62e4402b8b009699bb1f6578 : SUCCESS in 2m 56s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c93239c4f98045e3860e61a6de4113dd : SUCCESS in 1h 48m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c209275038abae8eec59f5af6f9a43668d193468","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-02-16 14:48:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"463e3fc6ec323bfc53dc5bcd60af661652f529ea","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-17 11:19:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ddf9f0d9d0246419464fdb1048dfe05801223ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:20:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da21103739c1e9dd4e8fb3ba6be395fc58c9e1df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 13:01:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8a0ad73dda842fb866fc2b2475d9b0d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d999eafb5ed452d8f13b95d83f19d9c : SUCCESS in 2m 43s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e289194b511b4e4da69352dc8a7ffa85 : SUCCESS in 3m 13s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8390026dc4044649806a1a2c621c5af : SUCCESS in 16m 19s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d83f21b2d8e247c797225cd04f646998 : SUCCESS in 11m 05s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c3ec42e5aea4e7aaffe4e8cb83ad41c : SUCCESS in 1h 10m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5034c45fb90844919dd2debb9050d6da : FAILURE in 56m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/438b5cd05d7c4dbcb72bca52989ed7c0 : SUCCESS in 15m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0cdf1143b1b48549957b981c6b574b6 : SUCCESS in 2m 52s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f25409393a25487a9922d1bcd8daac2b : FAILURE in 42m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9d44895ad86d6104fde83c5671b582daefc6719c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-20 01:38:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c3c4b199a7054274887ef12664fd65b8a360a516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 03:31:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9bb050778a94414be9b338bd935a85c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/618e74f676b542c783aa51f1b50307cf : SUCCESS in 2m 31s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2df075ce45a94b52b1554e8203d3cda0 : SUCCESS in 2m 31s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c0dfdf9fcf143aa9f1b1bf66f6dd34e : SUCCESS in 14m 51s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da8d2dad0dcc4045b24e6cd07f60f3c0 : SUCCESS in 10m 53s\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a574f3e4c46409eb30f857bc5834c16 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/614cb87371ae46518feed05b2f28bd3e : FAILURE in 26m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9660734a7da3485287ce27b08ce1954b : SUCCESS in 59m 21s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca6b704b3b4e491fa0b8f20e59fe8b40 : SUCCESS in 38m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cb2586f88274f1588aab3c99b704659 : SUCCESS in 49m 16s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f75a2b39926440ce907063c17af0ba8c : SUCCESS in 14m 31s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1956053e61d4d968e57f6054ab00bd9 : SUCCESS in 7m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6217dd46a30b4d37bdaa6560b9a90f2f : SUCCESS in 8m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd40e5c80124f7dbce452b5215df2c1 : SUCCESS in 2m 46s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b1c11c97f28c475e866118ac7068439a : SUCCESS in 1h 47m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b98b713149d8c0ce3bd6fe87edc727a9d1b062de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 03:33:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74b741e4e1e2483423e3cc4a8d1b51669c7d1217","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 05:40:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0b3466d52524383952398c64df5dc29\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b0857e36ea341b886bbaa92cc5e10a4 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61c5b4a784aa4903abe0bb6e02e7232a : SUCCESS in 2m 38s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/22dd517f6068492386740db9fa03599e : SUCCESS in 14m 47s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90e47d8380984b1f8e45d6e461125a89 : SUCCESS in 11m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8552af6fc8f46fbb53cd11e6d78c89f : SUCCESS in 1h 01m 05s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6f7df19abb2451c833f521a0e5c20c4 : SUCCESS in 47m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/3af1b3274c284c7285e82e2cadb1111b : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57bf1c9028094363a5c9befa6cbdc5c0 : SUCCESS in 2m 51s\n- puppet-horizon-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ecf6e93f0654bf38650b56e1c7e8643 : SUCCESS in 1h 59m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"701cc7283d86128c890a68157cd6fb8b59953d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 05:40:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a585fd099dc48e56cc279b40a81586d5da32173","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 05:41:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea9e9c428b3c4223bc357284e98cf2f3\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/591a17c5e17a480497533bbf77670f8b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f3e975a9e50048db4abd14c52f532298212f88a","revisions":{"4f3e975a9e50048db4abd14c52f532298212f88a":{"kind":"REWORK","_number":1,"created":"2023-02-16 03:08:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/874048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/48/874048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/48/874048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/48/874048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/48/874048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/48/874048/1"}}},"commit":{"parents":[{"commit":"77dcef9b78472a249ccb7c6667583590a2aed093","subject":"Add openstack_keystone_endpoint_type param","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/77dcef9b78472a249ccb7c6667583590a2aed093"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-16 03:07:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-02-16 03:08:14.000000000","tz":540},"subject":"Deprecate ineffective images_panel","message":"Deprecate ineffective images_panel\n\nThe images_panel key in HORIZON_CONFIG setting dict was once added\nduring Newton cycle but was removed when ANGULAR_FEATURES option was\nadded in the same cycle[1].\n\nBecause the parameter has have no effect for long, this deprecates\nthe parameter so that we can remove it in a future release instead of\nfixing it.\n\n[1] https://review.opendev.org/c/openstack/horizon/+/353377\n\nChange-Id: I376af3b7c4525036bf1bb449cb71490495395f2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4f3e975a9e50048db4abd14c52f532298212f88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4f3e975a9e50048db4abd14c52f532298212f88a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
