)]}'
{"id":"openstack%2Fpuppet-horizon~892483","triplet_id":"openstack%2Fpuppet-horizon~master~Id57cf05bcd30e8a575bd00d7e765d93cf87f9fef","project":"openstack/puppet-horizon","branch":"master","topic":"totp","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-22 14:39:30.000000000","reason":"Change was abandoned"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-10-22 14:39:30.000000000","reason":"Change was abandoned"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2023-10-22 14:39:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id57cf05bcd30e8a575bd00d7e765d93cf87f9fef","subject":"Add TOTP config support","status":"ABANDONED","created":"2023-08-23 08:17:46.000000000","updated":"2023-10-22 14:39:30.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"de45cb71fbc37a37c301839dda9b4a6077607df4","_number":892483,"virtual_id_number":892483,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-10 15:52:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 09:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 09:44:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-09-08 02:35:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"772173701e8158fb34bc3f09058e7c2d34d4b1f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-08-23 08:17:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa9d50cd548911f7b791067e82a9d811fad47f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 09:35:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ddd5f1eb86a47de8187581483e4d29a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51e91678b2564894b2a9c18d1ef450ae : SUCCESS in 2m 46s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/727423740ebf431c8780803b72b5f178 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/14b4bc2ba9ef4660a304c74762114d48 : SUCCESS in 5m 06s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cca97e98b84a4d94abf2ec422ed34f89 : SUCCESS in 12m 32s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec772d84ccbb4b67b94618e81eac9dcf : SUCCESS in 10m 45s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0a9c4dfaa1b4097b510d90ebe27e45d : SUCCESS in 12m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47cb0c97fbef47feb7b399dc9b593196 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d0c3814fd04e2da76114e25627539a : SUCCESS in 49m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8e5addea8494e6b8f392df9362e5b7a : SUCCESS in 1h 10m 39s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e7eb8ae06a84f20b3b11c74546e2708 : SUCCESS in 39m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05a3c824e5d44fa1b2ba7cde02c158b2 : SUCCESS in 57m 45s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22fa5dce5e894f58a5a2ddbdf109b545 : SUCCESS in 37m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa3ffac5caa74c09b6688fa813843379 : SUCCESS in 58m 41s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f1892d44dce47c4b7825e603e494f14 : SUCCESS in 14m 30s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/857022f96d6d4a63b3193556b01bac16 : SUCCESS in 8m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04c2493087764ecea575ae88c9629f8d : SUCCESS in 8m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/435b7f8b9cbe44f9835c395d86f6006f : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0e983e773afda84e2f5a4a5b012899db96bd379b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-23 09:44:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb15976dec5a262105200cfb04c72d5668c99f1c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 02:35:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"820b8c2eca4d5018fcb24895687f62948704f2ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-08 02:37:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74c05c6017c733a8d3fdb66abea1391107df0baf","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-10-10 14:42:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c9cec4a5c38e0438db2d164fa619cfe9a650962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-10-10 14:42:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2c8f7f56ea032a1b3378442e52247e88719b5ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:52:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d4e15b6d87d4bceacda28b5301aa1c2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b403ff1b68df45b6b6e7d2b0c8c506a4 : FAILURE in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/980fe7f972d442d7a17a1e528b283bb2 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b98f682775443089ad5e302a58f5e8 : SUCCESS in 5m 35s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1f72de648854f2196ea26efe80fd9bf : FAILURE in 8m 38s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2694e1e6c3b04affbeb60405f202f986 : FAILURE in 5m 14s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/226fb0ceccd0469fac91bbcafbba96f3 : FAILURE in 12m 33s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/670f16b584e94efaadb33b529cf13900 : FAILURE in 4m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c09288603ab464eb0ba359679c1e433 : FAILURE in 47m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/94a20b7d18b5465e8458688e8ce4e9c0 : SUCCESS in 1h 03m 59s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fec14a754d84f6eadbf63420951717d : FAILURE in 45m 20s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8f3d8c4963d4d5e95516fea46d43ea3 : SUCCESS in 55m 04s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbc2c801c8b74eb48194d686fa455154 : FAILURE in 48m 06s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/060adf4b4790458bafd15f5c3dcdd0c7 : SUCCESS in 1h 03m 06s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f948935d635541e38b3893edc87b643c : SUCCESS in 18m 39s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7588319b38c4dfa8c91b7da616bff53 : FAILURE in 7m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1647912da3f845ba867c521bba8a12b2 : SUCCESS in 9m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd2aefecfed843b98a463ca38afebc22 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"de45cb71fbc37a37c301839dda9b4a6077607df4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:39:30.000000000","message":"Abandoned\n\nI\u0027ll replace this by https://review.opendev.org/c/openstack/puppet-horizon/+/899001","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ead17812aa8ff3b26f916913321562be87d47e2a","revisions":{"38c04a9167890b8f69029560b3adefe1d12919fc":{"kind":"REWORK","_number":1,"created":"2023-08-23 08:17:46.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/83/892483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/83/892483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/1"}}},"commit":{"parents":[{"commit":"e399c8862a75555e343c2e58fd77993b830e710e","subject":"Exclude release note files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e399c8862a75555e343c2e58fd77993b830e710e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-08-23 08:16:39.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-08-23 08:16:39.000000000","tz":120},"subject":"Add TOTP config support","message":"Add TOTP config support\n\nIn this patch:\nhttps://review.opendev.org/c/openstack/horizon/+/885570\n\nHorizon has added TOTP support. This patch adds support for it\nin puppet-openstack.\n\nChange-Id: Id57cf05bcd30e8a575bd00d7e765d93cf87f9fef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/38c04a9167890b8f69029560b3adefe1d12919fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/38c04a9167890b8f69029560b3adefe1d12919fc"}]},"branch":"refs/heads/master"},"ead17812aa8ff3b26f916913321562be87d47e2a":{"kind":"REWORK","_number":2,"created":"2023-10-10 14:42:14.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/83/892483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/83/892483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/83/892483/2"}}},"commit":{"parents":[{"commit":"e399c8862a75555e343c2e58fd77993b830e710e","subject":"Exclude release note files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e399c8862a75555e343c2e58fd77993b830e710e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-08-23 08:16:39.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-10-10 14:41:50.000000000","tz":120},"subject":"Add TOTP config support","message":"Add TOTP config support\n\nIn this patch:\nhttps://review.opendev.org/c/openstack/horizon/+/885570\n\nHorizon has added TOTP support. This patch adds support for it\nin puppet-openstack.\n\nChange-Id: Id57cf05bcd30e8a575bd00d7e765d93cf87f9fef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ead17812aa8ff3b26f916913321562be87d47e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/ead17812aa8ff3b26f916913321562be87d47e2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
