)]}'
{"id":"openstack%2Fpuppet-horizon~899002","triplet_id":"openstack%2Fpuppet-horizon~master~I61aefe65e6218a79dce7a0633cb0ceb13b0021b4","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-17 09:18:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I61aefe65e6218a79dce7a0633cb0ceb13b0021b4","subject":"Require valid type for available_themses","status":"MERGED","created":"2023-10-22 13:03:38.000000000","updated":"2024-04-17 09:18:00.000000000","submitted":"2024-04-17 09:18:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899002","meta_rev_id":"c031a73083572b14b29f827df06e9319dda21354","_number":899002,"virtual_id_number":899002,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 09:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-01-12 16:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-04-17 07:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-04-17 07:53:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-22 13:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 08:07:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-01-12 16:31:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d461b7b75d6e3945473dffb18e807ec4f1933fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 13:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aad7fe29334c23130d110e0f179c24a71b7ae14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-22 13:18:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48e7422a3d354fe097cd5acc26305073\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b30cdec5f0264550a021704b349ca5d1 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/328cb54025734d85a6bd9430e08e42a9 : FAILURE in 2m 34s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd50056140334d3cb023feb60074ad45 : FAILURE in 5m 40s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a627784c8cb40c783196ef27c8290ad : FAILURE in 8m 35s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85c2aec0ed4f426f8b07127e6faf306c : FAILURE in 5m 35s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/0abd6ef5ab42435b832bb085975b3d42 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a9c09e24c943b18862d70028639120 : FAILURE in 5m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7742e462da6e4d5d8304a8145c630419 : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/33cbc67c413f47569e520e36ed45e516 : FAILURE in 11m 24s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd489da4c4f24195b959191e54186b5f : FAILURE in 6m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b36bbb3450742ba97875aab017e66ab : FAILURE in 9m 39s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/094042d7f81a4dd08196f5d39cbca98b : FAILURE in 7m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/705a6f990f3440098e67f9894043f948 : FAILURE in 8m 57s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f10bb3bd18949f09ffb6926ef6d2ca4 : FAILURE in 9m 12s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1bbfc283b29491d8026e68fb2d85514 : FAILURE in 5m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b56010c97c04ed4a8d6ac391e7d1e0a : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3610b1f5d39a5c2a12f3d9e0e5a718563d813d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 13:44:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"02b5a3ba60756f333fc2c6a691fb606374f23888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 13:59:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8277d326cf1191323593e52e5b9cb637d4aedd8d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 13:59:15.000000000","message":"Topic totp removed","accounts_in_message":[],"_revision_number":3},{"id":"dd4673c96a66bc8520708511cf24c62477349011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:00:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bf7bf5b523759e020bbb7cb081833622337e6be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:01:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"80b82c0ae4d6e96770ea8ff8451652d89d1e6ce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-22 14:19:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce078654a1804d7e8527b7abaf189cbd\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfc65a4e94ac45ddba7abe09838a5d62 : FAILURE in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd60a8fefc10448ebc2091ea9359c0df : FAILURE in 2m 19s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d320c7fb9394b16a65e9eb43327cebf : FAILURE in 5m 41s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/97eaa361b36449a995601b49bb8a3c7d : FAILURE in 8m 40s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a8101956eb14e8a9f9f429c6ae31d23 : FAILURE in 5m 05s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d436b061c51848918436dc1223057b3e : FAILURE in 7m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/427db95fd4a34f18938cf88b9f75eb07 : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5842aa7a8341421e9f5bf8de526ba61a : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1994a795b794d6b9966c3733431dd64 : FAILURE in 11m 45s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b63287c2b78d419bb9a0faf79ac16495 : FAILURE in 5m 18s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c849475a3c43438205abb32aea8b39 : FAILURE in 11m 50s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3effc88c538c4027bdbe1e921c60ab35 : FAILURE in 5m 37s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/635c7693e86a41beafa3579cd52abea2 : FAILURE in 11m 55s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd990513eb76429087c1508f59f02b78 : FAILURE in 7m 20s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c358303c448f43449c1de61da8d00c35 : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/257bf69102a34a568989f67900e51ff8 : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"36f6b22e892f16f1c0e09e849bbee1032a89b12b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:42:00.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0a9b4891589015f928e8c2e08696aa0851295e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:42:06.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"aadd28e79911a60796f282e063dc2cf023aa8eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-22 14:52:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"accbe9c7baccadcd8b202c85c55e6c06f3892002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-22 16:01:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d11d3e2735774b3198f3417100d0d1b4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f1118deee654dd182c7d39ae856696b : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a9a422df5e549f7a9a20a4278d010ff : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/65ef3f160445441783e31507b682ad62 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e0d605052ce4ff78e2095df8cf22336 : FAILURE in 14m 30s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65d3b0a10e6f4f3799ffcd744346cf84 : FAILURE in 11m 27s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/f617f4ba28a14492857fd1bf8e94051e : FAILURE in 15m 03s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/968a939f78044f7ab29408e6394d000e : FAILURE in 11m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/439b237cd4c74db39199cf868ea018d3 : SUCCESS in 53m 03s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5a6f9f03e714badafda31bf7e8f5d21 : SUCCESS in 1h 04m 02s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c579b62645141b7aad8c1274311c75f : SUCCESS in 45m 00s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b78a4f2e01341c08c170bc86366f7d0 : SUCCESS in 56m 17s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3398f374de84b7dbc71ef3fda109a9c : SUCCESS in 35m 23s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d198a0eb5e4d1783e7554afa157620 : SUCCESS in 43m 01s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/8fe76564da6b48919d106261ada2ee52 : SUCCESS in 11m 17s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56fe7e95d7224336a2c804f9bf8da233 : SUCCESS in 7m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06d45c4c174440e783baeb65a2f30748 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"661cd2be424e6c7cce4607b1b353ad4636c09fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 06:30:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"2dd5370093c58c0c159350a9ce532d99b7d1b4d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 10:02:47.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"eec9bc5e8ed9f7b91288c69147d67c766497238f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 10:22:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/983a508880784b9ca379e67cb8236dd5\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4bb71b1776846aa8edec27ace62965f : FAILURE in 2m 30s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68a44688de4f49d9bb9dc4d5484ec978 : FAILURE in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ade9d7f81cb846ce912f55f5b1e35eb5 : FAILURE in 5m 23s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3e25903f217497fa16bb358b9850a37 : FAILURE in 12m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b20abba57304aaf8b633078ab893ddd : FAILURE in 7m 40s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf896b3f58f34aefb3c997041b72ed7f : FAILURE in 12m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07ea88ae4ff448ee887e0cf114f04bf0 : FAILURE in 8m 44s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b155fffcbbbe484cac2fadd5b2a2e10a : FAILURE in 6m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00298dbfae944bb5a39c58c512aefffb : FAILURE in 11m 15s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5befc79a07c4d44a36515e1128a5508 : FAILURE in 5m 55s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2805a3e17bd145d886625ff9eba38fa0 : FAILURE in 11m 55s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02c888da98844ce69290a4ba76b713a8 : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6920f4604bc3434ebc09a47652962cc4 : FAILURE in 8m 59s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2855eb47e40b40b08f9402fc8471bba7 : FAILURE in 8m 06s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/473a981382f942298550b60ecd07d6df : FAILURE in 6m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbb37b8121814270af51c6c7aea4f771 : FAILURE in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"580d8c0725cdd2c6ede0d5e1d8d3d8b29055592f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 05:06:55.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1c9b98a72cc4ad1732f5b5501869d5b3114cd1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 05:21:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b78468992434b3d942618af1faa546d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6083dd527b0f41ea91200a3f1e51013e : SUCCESS in 2m 20s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f7091653499416f987403b6368b833c : FAILURE in 2m 58s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b0563b06392468ca828a74d5ad0d54c : FAILURE in 5m 05s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/04ab02bd58b04454ad48984fd9a62a49 : FAILURE in 9m 08s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff2e1f606de84faaa9aa8f0acb78fea6 : FAILURE in 9m 04s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/57ce191d82e647c5870b9eaaadf0b09c : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04de262a5acc4a0e9c6e8cb5a61c0ed2 : FAILURE in 9m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb60259f570e456cb44384e29c2fe951 : FAILURE in 6m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7134e1f16164bf3ba9d997b731c6c6d : FAILURE in 9m 16s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6533d19c0054416982cb797dda5a7e0 : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c4733ab314a4496b8b5a8302f07c3f7 : FAILURE in 8m 58s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a0dfd865eb44830b520464c6429716c : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5872595822484207bbde700e48078fb6 : FAILURE in 11m 10s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee0a1ae3852c483eab36fa34e3fad550 : FAILURE in 7m 56s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/052b404146dc4acfb557d7492f257292 : FAILURE in 6m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b910ce324364220a459009ee784ba04 : FAILURE in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"589c42823e8d22b078a2aaa6c14124ee8ec25380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 05:41:21.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b557a0703ebbd3f486f0ce23fa2266ccc773d530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 06:45:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e10ca9dea8f242f4954b3b6547e95a0c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/350fdd72bfd64b868739892bdba05f73 : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef60b395cd6545fd8750ef0a83dabf76 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bf67a7573e94434952967847df721d3 : SUCCESS in 5m 33s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbd8b16587654156a5d0c1ce5bbf1696 : FAILURE in 11m 07s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a1617e262724840b10b8210b86e66ef : FAILURE in 11m 32s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/2382aa9dd5f54b6da60b739aed38addf : FAILURE in 11m 24s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6059b98618d1436faa60185235fe8487 : FAILURE in 9m 16s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5ed389cbf3448a3973098bbbc4aebdd : SUCCESS in 51m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9fb8265be1140d1b8e43a297ddf21e7 : SUCCESS in 49m 54s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4739b6013384d2aad61cad2e903c2fd : SUCCESS in 36m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44e04a79615c480ab648991d7363bc53 : SUCCESS in 58m 14s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24ccb6728d044aa591d0974d19559fe5 : SUCCESS in 48m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e95435942c2945ff98b494bd8b9c2daa : SUCCESS in 56m 44s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d0ecda7a1eb4fdba3c01b2d5bf957da : SUCCESS in 14m 27s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ea576b17d82455697b9d5ab6f451bbd : SUCCESS in 8m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/168b753d883b4b93aa4dc31e42e047d7 : SUCCESS in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d43df5b4ce4c4321763a4a944f454338ebfd2a85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 07:13:39.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"74338c04bf731048831de84ae427b3356997f24d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 07:28:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"15846bef457cd3e7337171faf6149a60175334e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 07:49:11.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f647f53e47b4a5aa5a37148dcfc6f90\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f27e4ba63be94f5ebc6ee785dfcddba0 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e6ea459227b4c03907f43a48eaf7670 : FAILURE in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f259330ac6f44dba960d8a00b53113e8 : FAILURE in 4m 42s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/23dd5a76f06342889b213c4bd8ba560b : FAILURE in 12m 05s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca648fc1aff5454cac8b473c508d5586 : FAILURE in 9m 07s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/054ce06a86e44259a98ef61b44c52742 : FAILURE in 13m 16s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13a6e85c08c54f9aa64afedb9308253e : FAILURE in 7m 41s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/133a5cd2ec7a4222a93609e1a4e1c1e8 : FAILURE in 7m 02s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/55f055b7a9384215924ef6383d0f98e9 : FAILURE in 11m 46s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9616d1bfa06f46d48bbf1f23715e14c6 : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3f0567607f346459313a057fee62480 : FAILURE in 10m 07s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c301135e46d46018407dc66b0bed695 : FAILURE in 7m 10s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be389dd6ee244a129de2973b8305c9a0 : FAILURE in 11m 56s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/59397871b73d4fa4aa4529b9bbba1d97 : FAILURE in 9m 32s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/405b156a171649b09adf8eef12469854 : FAILURE in 5m 48s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c7b05d0dffd41888289fd59e16ae571 : FAILURE in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f5136aaf73a4a986d31c04c075497aa2557f6c33","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:07:36.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"ccc829576c0cb7bfb75e86b1ea1280a65baa586b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:07:40.000000000","message":"Patch Set 14: -Code-Review","accounts_in_message":[],"_revision_number":14},{"id":"bc806e5aeef7547ee07c1492dc812864c8c814d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-30 15:56:51.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"bb5ee44bf0524a4ae12def9b3c651f546f3de58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 17:39:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b543022ed9f3430b9ca6224b7626871a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4efb8a68071b44d28fbc01640e99d9a4 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e42c01761c741409bc1e380b65f7611 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c905187c84dc49dcab782f725d337c1f : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2bb5c3bf1d94aedaa880c012a06a4af : FAILURE in 8m 19s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/103bdc91c253427285716ea9ebeb7b84 : FAILURE in 5m 26s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/537b99f7459642c6a80764b4a6e25dcc : FAILURE in 8m 02s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6550c10af20841918fec6274af3abba0 : FAILURE in 7m 30s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f09012552a7427ab5c14b996091bb4a : SUCCESS in 47m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03d5fcb46ae14fdab05b1c551ee3ee17 : SUCCESS in 1h 08m 49s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7bbed234f074cbb90fd8c4b14f9ae00 : SUCCESS in 40m 46s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e839dd61c0249c9884b76a0b55f5ef2 : SUCCESS in 50m 52s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c7f3d7a84784d779103e7ad25759f1e : SUCCESS in 37m 36s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af4901570af448deb011cf753a614f9c : SUCCESS in 53m 46s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2f3f353660b40d0b9167f42d775f532 : SUCCESS in 12m 54s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74f054055da0451b8e995f7432b8210b : SUCCESS in 8m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c5caf4ce5454befa9870228d5e0dd96 : SUCCESS in 8m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e7190e21731364566841ba3842da418f58a1323c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-09 17:02:11.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"8c7a5a77286e8817f28df03661c19df805f8f93a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 18:33:24.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9a5b162ac144cb1a36746f4047df3ec\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6833ec3db9e544349cf3c52761636fa5 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/583ae1ed6aca4a4093cfb41af2b8b381 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f0f235085ee4c27b16c22d7c0f8cf1b : SUCCESS in 7m 07s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/90a61e83572646fd8d0a05491c8c3ae6 : FAILURE in 15m 32s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/594ba6cc07b04cefa6e0feb317f8e092 : FAILURE in 12m 35s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca1cd1ed2d4d47e98a16b0ffdb3ad2ec : FAILURE in 16m 52s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78b081123d6a46df99fa4378327302f5 : FAILURE in 11m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97ac6739c814fa3a143f4d2c6b41329 : SUCCESS in 56m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea0e3a82715d4a59bad8bdd611d7233e : SUCCESS in 1h 22m 36s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a50ed3d855a4833a1e99b0bcc11d774 : SUCCESS in 59m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/687fe7941e394b5a9717c25fa406ecad : SUCCESS in 1h 07m 35s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6febdbe3bf24600a861a8232e468a4c : SUCCESS in 43m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/07f6e780a1ae4d809fefbbeba86d79f0 : FAILURE in 1h 06m 58s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1ecec4830694da091028c0c448ed8dd : SUCCESS in 46m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aba5c23f310341758380c3c273a6409f : FAILURE in 51m 44s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/a019a2101af147d8b665e61985c9348d : SUCCESS in 13m 27s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/229168cea7024c909a0cd009842c375a : SUCCESS in 7m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f61208287524883ade8d41b49158cbc : SUCCESS in 8m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ededed990a4339911a5c2b24763603e5dec06e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 10:32:35.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"a1aa7df1ddd0e738352d63f7f16aacde1d8407d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 10:32:56.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"63c796cedfafe23d88991e6e2be43497c6503a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 12:18:20.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a60c0723b7b24fa4bdc0f29ca56168c8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d716e421c5ea4c588df5dd7622852692 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91918fcb68b9425085cb4b63add7cdae : SUCCESS in 2m 46s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1c2c61d7e5f4d729a1af0b0ae899605 : SUCCESS in 6m 20s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e54fa8065e74a0a970237f0d261d6e4 : FAILURE in 18m 02s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da98666a99b74294897ca0766ee26c7a : FAILURE in 12m 26s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/64ee6c7da74b405880a126327cb7b6dd : FAILURE in 15m 19s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a369b3bc239492583a775b453f75a56 : FAILURE in 11m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e94464cb7da04c42ba1338fa1fd070e7 : SUCCESS in 1h 06m 28s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c9c989cefff480896c714656e37a1e0 : SUCCESS in 1h 14m 38s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe4dcc85a8e846e6ba1cbc4fdbf6f911 : SUCCESS in 56m 00s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d5c7fca30e4d1197a1f6716014aafb : SUCCESS in 1h 23m 31s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/201281d95f4d437682a19d1874574533 : SUCCESS in 53m 06s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be02d9ce4ff64df3b89c29288ae9b812 : SUCCESS in 1h 00m 04s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b599c5a85ec43e5b496c92c6c51223f : SUCCESS in 50m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e0cb36d95f6499796824786fffee9ed : SUCCESS in 51m 25s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d54218707714f84ae765c9f235bb3b5 : SUCCESS in 13m 58s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/955a97aed34e4a839c0154adfb412579 : SUCCESS in 9m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f99717cc7c94b47b39d641a3e5fcddc : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"bfa52b7044581aadc7620f947ba791cd6d747476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 12:56:08.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"38e4673533aa948421296e68cb88122f927ccfd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 12:56:57.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"4285abb34be74840c32b667e711f9d7e9008b6f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 14:23:38.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5b4b341d0164a6585011ccf06b6fdf7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f66460261ce54d06a8dd30a618eeac27 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25bd90bfd5c04ebbbc1da33f079160a8 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/204c83cf2c76479f91d83ffae8db75c0 : SUCCESS in 5m 54s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/bc6d3a7a5a29483ca651b82132ac1261 : FAILURE in 11m 35s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4019c47b30df4b108c7c51a351746942 : FAILURE in 11m 13s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6a3e73accd74eef9b6cfa7694134c07 : FAILURE in 12m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64a259755e7d4ba5b0c6b9f7c62d4cf8 : FAILURE in 11m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/239223da773b49cc9c9cdf63d99ad610 : SUCCESS in 56m 20s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b9b3a3733b740e6a160567cdf3953d7 : SUCCESS in 1h 16m 56s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89572775eecf45ce8d5ee65d37837805 : SUCCESS in 1h 02m 54s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0fb35013f744783a6f17b71a4ea8c32 : SUCCESS in 1h 05m 42s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b73942ecf2240bbbf9acc09b759c002 : SUCCESS in 46m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/717e1f27eb20448d895d54b1e46bbf0b : SUCCESS in 54m 09s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ab0694235546efac811a373371f057 : SUCCESS in 41m 35s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/676316343c7d4c6182ccb0f73b4b62e9 : FAILURE in 1h 04m 28s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/9bf8d7aa5abd4865b310f7a928850ae1 : SUCCESS in 13m 33s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f601c5b325cd4b61a2acbf9ec3b2dc6e : SUCCESS in 8m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbf5d47f30c449bfb5b4d8d66a21284b : SUCCESS in 8m 35s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e03cb8b812e058f7e12312809820fa8d64c28d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 14:54:31.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"9731a34cbb01c4e6c522efd8b0d1e7e825ab0090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-10 15:42:03.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"d080c416aef3fa04b78a27ab1eecc7f4e5f1b786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 17:19:34.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/454a9316c64f4e7f80a96092da138331\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b229120b9b7e43109b381a4eac284d89 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ecd14e362ff43448050a93a980435c0 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/7440040922d34084aeccdf4d064270a1 : SUCCESS in 6m 32s (non-voting)\n- puppet-openstack-unit-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/57a1bae413fa41299da463bf0b80ad65 : SUCCESS in 12m 04s\n- puppet-openstack-unit-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e70802e9dda84dbfa70aa3cbe3920238 : SUCCESS in 11m 58s\n- puppet-openstack-unit-8.3-centos-9-stream https://zuul.opendev.org/t/openstack/build/09e16533a3c641d99798dccfdc951728 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e32b76b0de524639beef9da03c8ada57 : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/344f2c640d3b481c86778d2b272d490b : SUCCESS in 57m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7ad7744ff1a45718c4ff3565db623e3 : SUCCESS in 1h 29m 46s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/386c3d9d6cb946cfb0de5564f8ce79ba : SUCCESS in 55m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/28efeaf641c247d6bb0a842087bca560 : SUCCESS in 1h 27m 48s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2854653da9a44b2f9de1286a6832395d : SUCCESS in 45m 36s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5655ab082a04499fa603b43aef3f8b58 : SUCCESS in 1h 06m 29s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c8765b353d4cc9aac81c390c1a7e42 : SUCCESS in 45m 37s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaddc10e14b44da0a41095f35e4def6c : SUCCESS in 54m 19s\n- puppet-openstack-litmus-7.27-centos-9-stream https://zuul.opendev.org/t/openstack/build/337946f5b8ba46668c4a84ee6196cdae : SUCCESS in 17m 04s\n- puppet-openstack-litmus-7.27-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d4a8fe67be49b2b8100884a4b11005 : SUCCESS in 8m 46s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed90b66ed17449348aa5455333a2defd : SUCCESS in 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0a315b27ca022bcf44717f4bcf75d2b3c5bf8058","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-01-12 16:31:37.000000000","message":"Patch Set 22: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"db8494eceb2771d318851df3d844b632784e1d8c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-17 07:53:20.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"bd441e3470b86a4cc39f269a6f8b3a37922f9712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 07:54:14.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"72068db5a9dbdf3e40b05f6d6dd51c6fadb310cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 09:18:00.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5eda581eb64113a226e0d2d79c7aa1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52e8ec51512846d48813c5d154168f4e : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee691669219e4fd6a09a68ae37bff62c : SUCCESS in 2m 15s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/64d0a5d379bc4ebcb8e83ef66277a5ed : SUCCESS in 13m 19s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5d192a616f04f6a83825f4438476603 : SUCCESS in 9m 20s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d085f03ea4ff4581a3243f0f9eeab43d : SUCCESS in 1h 18m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3128dfecd39947809836a638fcbacce4 : SUCCESS in 51m 20s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/304d9805e74b401093522171cc8e169a : SUCCESS in 56m 04s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e6c2261c99b4dea9d55fb57885a1119 : SUCCESS in 57m 43s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f915660158c4641b56cac61db05d152 : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":22},{"id":"c031a73083572b14b29f827df06e9319dda21354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 09:18:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"5a3245f9d494ca864e73b130e1413c3edc44f50a","revisions":{"b0a399e99c3ebcfdfb05ba8fdb25369376196725":{"kind":"REWORK","_number":1,"created":"2023-10-22 13:03:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/1"}}},"commit":{"parents":[{"commit":"6feed0622813837b60cc3808734909dcee5208c3","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6feed0622813837b60cc3808734909dcee5208c3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results.\n\nThis enforces an expected parameter type instead.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/b0a399e99c3ebcfdfb05ba8fdb25369376196725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/b0a399e99c3ebcfdfb05ba8fdb25369376196725"}]},"branch":"refs/heads/master"},"bd92149546832f05a740f4d53fe4e19b220fdbae":{"kind":"REWORK","_number":2,"created":"2023-10-22 13:44:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/2"}}},"commit":{"parents":[{"commit":"eec1c15570e450db3e7423089ae8f27c59ebe88d","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/eec1c15570e450db3e7423089ae8f27c59ebe88d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:44:42.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results.\n\nThis enforces an expected parameter type instead.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd92149546832f05a740f4d53fe4e19b220fdbae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd92149546832f05a740f4d53fe4e19b220fdbae"}]},"branch":"refs/heads/master"},"4ce155e25b2a10b26fd93d5e7017099a7d1b8f50":{"kind":"REWORK","_number":3,"created":"2023-10-22 13:59:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/3"}}},"commit":{"parents":[{"commit":"eec1c15570e450db3e7423089ae8f27c59ebe88d","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/eec1c15570e450db3e7423089ae8f27c59ebe88d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:57:51.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4ce155e25b2a10b26fd93d5e7017099a7d1b8f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/4ce155e25b2a10b26fd93d5e7017099a7d1b8f50"}]},"branch":"refs/heads/master"},"d03a16a6730f79d08a8850a33a3017c3548e6e50":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-22 14:00:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/4"}}},"commit":{"parents":[{"commit":"bd3a244127db257ca017f74880fe960576abc2e4","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd3a244127db257ca017f74880fe960576abc2e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 14:00:15.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d03a16a6730f79d08a8850a33a3017c3548e6e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d03a16a6730f79d08a8850a33a3017c3548e6e50"}]},"branch":"refs/heads/master","description":"Rebase"},"c4bbd4f9165a9035309cb3398c5b3563655df748":{"kind":"REWORK","_number":5,"created":"2023-10-22 14:01:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/5"}}},"commit":{"parents":[{"commit":"bd3a244127db257ca017f74880fe960576abc2e4","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd3a244127db257ca017f74880fe960576abc2e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 14:01:30.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c4bbd4f9165a9035309cb3398c5b3563655df748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c4bbd4f9165a9035309cb3398c5b3563655df748"}]},"branch":"refs/heads/master"},"1902bc1d4d58c47bd2d43dae393173e28e6c5ebe":{"kind":"REWORK","_number":6,"created":"2023-10-22 14:42:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/6"}}},"commit":{"parents":[{"commit":"bd3a244127db257ca017f74880fe960576abc2e4","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd3a244127db257ca017f74880fe960576abc2e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 14:42:00.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1902bc1d4d58c47bd2d43dae393173e28e6c5ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1902bc1d4d58c47bd2d43dae393173e28e6c5ebe"}]},"branch":"refs/heads/master"},"c07d0970eab9a568e28364d24155eb5667dc6218":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-10-22 14:42:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/7"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 14:42:06.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c07d0970eab9a568e28364d24155eb5667dc6218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/c07d0970eab9a568e28364d24155eb5667dc6218"}]},"branch":"refs/heads/master","description":"Rebase"},"7f9bf6eebb31664b91f2417be04d70201eaee025":{"kind":"REWORK","_number":8,"created":"2023-10-22 14:52:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/8"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 14:52:06.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7f9bf6eebb31664b91f2417be04d70201eaee025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7f9bf6eebb31664b91f2417be04d70201eaee025"}]},"branch":"refs/heads/master"},"a63fd41774bb51408f9242d85d409fc44ba3f14f":{"kind":"REWORK","_number":9,"created":"2023-10-23 06:30:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/9"}}},"commit":{"parents":[{"commit":"bd3a244127db257ca017f74880fe960576abc2e4","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/bd3a244127db257ca017f74880fe960576abc2e4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 03:21:20.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/a63fd41774bb51408f9242d85d409fc44ba3f14f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/a63fd41774bb51408f9242d85d409fc44ba3f14f"}]},"branch":"refs/heads/master"},"a0058afd90d35fedc06441467142a9dd4c05e408":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-10-23 10:02:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/10"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 10:02:47.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/a0058afd90d35fedc06441467142a9dd4c05e408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/a0058afd90d35fedc06441467142a9dd4c05e408"}]},"branch":"refs/heads/master","description":"Rebase"},"8bad7faa9beafcf45001ce11b226028dbc2ced7a":{"kind":"REWORK","_number":11,"created":"2023-10-26 05:06:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/11"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-26 05:06:12.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8bad7faa9beafcf45001ce11b226028dbc2ced7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/8bad7faa9beafcf45001ce11b226028dbc2ced7a"}]},"branch":"refs/heads/master"},"d4c41ef8f8cffea3a589c474dc6b2d47bf681b05":{"kind":"REWORK","_number":12,"created":"2023-10-26 05:41:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/12"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-26 05:40:39.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d4c41ef8f8cffea3a589c474dc6b2d47bf681b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d4c41ef8f8cffea3a589c474dc6b2d47bf681b05"}]},"branch":"refs/heads/master"},"28a2e3310e0c804fc9110865cbc57fd347e2d7eb":{"kind":"REWORK","_number":13,"created":"2023-10-26 07:13:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/13"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-26 07:13:09.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/28a2e3310e0c804fc9110865cbc57fd347e2d7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/28a2e3310e0c804fc9110865cbc57fd347e2d7eb"}]},"branch":"refs/heads/master"},"fd9b35e173b5ec096f1ffab0a144c2bb3f73b78f":{"kind":"REWORK","_number":14,"created":"2023-10-26 07:28:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/14"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-26 07:27:03.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/fd9b35e173b5ec096f1ffab0a144c2bb3f73b78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/fd9b35e173b5ec096f1ffab0a144c2bb3f73b78f"}]},"branch":"refs/heads/master"},"1322f465597ef300cf3cdb884c02825f309b2c25":{"kind":"REWORK","_number":15,"created":"2023-10-30 15:56:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/15"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-30 15:56:43.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1322f465597ef300cf3cdb884c02825f309b2c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1322f465597ef300cf3cdb884c02825f309b2c25"}]},"branch":"refs/heads/master"},"03756eb27f8cd3c20549528808db816a8b582c3a":{"kind":"REWORK","_number":16,"created":"2024-01-09 17:02:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/16"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-09 17:02:11.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/03756eb27f8cd3c20549528808db816a8b582c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/03756eb27f8cd3c20549528808db816a8b582c3a"}]},"branch":"refs/heads/master"},"f83ecf893f734d2a717accd8d41ff9c2e881a212":{"kind":"REWORK","_number":17,"created":"2024-01-10 10:32:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/17"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 10:32:34.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f83ecf893f734d2a717accd8d41ff9c2e881a212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f83ecf893f734d2a717accd8d41ff9c2e881a212"}]},"branch":"refs/heads/master"},"865a481365680948730eca50c829bc4ba373baba":{"kind":"REWORK","_number":18,"created":"2024-01-10 10:32:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/18"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 10:32:55.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/865a481365680948730eca50c829bc4ba373baba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/865a481365680948730eca50c829bc4ba373baba"}]},"branch":"refs/heads/master"},"64ba232f294076349d9b1c184e053fe8914a673c":{"kind":"REWORK","_number":19,"created":"2024-01-10 12:56:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/19"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 12:56:07.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/64ba232f294076349d9b1c184e053fe8914a673c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/64ba232f294076349d9b1c184e053fe8914a673c"}]},"branch":"refs/heads/master"},"d9a599b98edd52014ffec9336ba0ec7237b0c93c":{"kind":"REWORK","_number":20,"created":"2024-01-10 12:56:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/20"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 12:56:56.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d9a599b98edd52014ffec9336ba0ec7237b0c93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d9a599b98edd52014ffec9336ba0ec7237b0c93c"}]},"branch":"refs/heads/master"},"1009d7a758fd3c7347dfbab5347e2488000dc781":{"kind":"REWORK","_number":21,"created":"2024-01-10 14:54:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/21"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 14:54:05.000000000","tz":540},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1009d7a758fd3c7347dfbab5347e2488000dc781"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1009d7a758fd3c7347dfbab5347e2488000dc781"}]},"branch":"refs/heads/master"},"5a3245f9d494ca864e73b130e1413c3edc44f50a":{"kind":"REWORK","_number":22,"created":"2024-01-10 15:42:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/899002/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/02/899002/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/02/899002/22"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-22 13:00:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-10 15:42:02.000000000","tz":0},"subject":"Require valid type for available_themses","message":"Require valid type for available_themses\n\nCurrently, if a users gives invalid input type for available_themses,\nthe logic ignores the input and reflect nothing in the config file,\nwhich can cause unexpected results. This ensures the given input is\na hash value.\n\nThis also adds validations to ensure only required values are included\nby the hash.\n\nChange-Id: I61aefe65e6218a79dce7a0633cb0ceb13b0021b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5a3245f9d494ca864e73b130e1413c3edc44f50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/5a3245f9d494ca864e73b130e1413c3edc44f50a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
