)]}'
{"id":"openstack%2Fpuppet-horizon~900839","triplet_id":"openstack%2Fpuppet-horizon~master~I7e074f78b15ab22d86d6a6a10f573cfc4052023a","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-16 02:47:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7e074f78b15ab22d86d6a6a10f573cfc4052023a","subject":"Support purging local_settings.d","status":"MERGED","created":"2023-11-14 01:31:58.000000000","updated":"2023-11-16 05:12:42.000000000","submitted":"2023-11-16 05:11:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900839","meta_rev_id":"4b8741f1dee66319910e351e615dfaccbb4fb639","_number":900839,"virtual_id_number":900839,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 05:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-14 16:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-15 15:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-15 15:40:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 04:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 16:48:30.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-15 15:40:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2bea1b0ebbd016a2874b1deea2fe85b4f2619873","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 01:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9cf8820e71cfecf090d714dcc369baba7101c82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 02:16:09.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d8664255b680e0bc4f2504af0eb27efaa86f288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 04:20:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9f4c768746048d59ca295a0e7b5e8ef\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de73e157722646958209278981d5be40 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/957c199944c244f1859b6d8fdc9e509b : SUCCESS in 2m 56s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/295bc9b92b45406680bbbdc590cf96df : SUCCESS in 6m 14s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ee58caeeaae44e78d8b66cd099f5122 : FAILURE in 17m 38s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15a4dc528548476face92a8cbb7af7ae : FAILURE in 12m 15s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d64af7d389742da949edd9f3d9596d0 : FAILURE in 14m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a0815cd0a184a888ba838af57dd6e3e : FAILURE in 11m 33s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d481fa80e1a438880aaa35213aa3810 : SUCCESS in 50m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c023713220e74ed4b35f12c769119b2c : SUCCESS in 1h 03m 46s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fef278aa0ad4b90b62e8cd906334a44 : SUCCESS in 50m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a875786bdfd4409bcb535f5adb27100 : SUCCESS in 54m 21s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2486127f96043bfa8635e143f362068 : SUCCESS in 45m 00s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1631c2cba9f487ab9ebfe00bc65dc31 : SUCCESS in 54m 45s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c63197e1f96b42c7b0f81eb3c2733487 : SUCCESS in 15m 26s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b96a862779f54a5cbdc7e854e50d10a9 : SUCCESS in 11m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/adcf263a08cc4d5abbaa856908d2247c : SUCCESS in 8m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bb8ea587f6a4972a442396ffbfaaaa3 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b7a4a7f1227a8cdaa0e1333b2664afac92599dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 08:25:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"37cb58306cae43129fbf73572d18656d138f98e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:34:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a254af2f7e6443a6b7d47033f0c609f8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e03e129f712240f5a5b96a7ca2b821d4 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f3ee0921a7c42f986aa094bfeb90077 : SUCCESS in 2m 35s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d98b6b8ed54a49c0a9f4eb77b8fac509 : SUCCESS in 5m 41s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ed48f154c65456f848121dac44df89f : SUCCESS in 15m 57s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f893bcfe6654eb29bccd4724fae0eea : SUCCESS in 12m 26s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/068dd6258eef4e5da2593164e82df959 : SUCCESS in 15m 05s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7cd2cf18f5a4849a41ecea087f1c495 : SUCCESS in 11m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/678a80499c07453eb00c6cba2362d875 : SUCCESS in 44m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1194de15e1ae4701a65e8f4d40be1660 : SUCCESS in 1h 07m 13s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec39d4f67fb44ba4964f9c0114d35a05 : SUCCESS in 41m 26s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c250988aac840c085615e202c9609f4 : SUCCESS in 1h 01m 16s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02a876d8488d42f0bc6f3aa69ffa49ea : SUCCESS in 41m 08s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d5145987a9d483c9be9225b2bd4e087 : SUCCESS in 47m 18s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c816389e0ed94fa2b46e73b35b83e9cd : SUCCESS in 14m 49s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4198c502dc794109976d59043e2be9b5 : SUCCESS in 8m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1847a6efd0b6424dacae720379468bc4 : SUCCESS in 8m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f87399c8a3534ea69d5d367caf9eb822 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"fcdb6359243bcfbaf13e3e23eb690b9fe88600a2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-14 16:48:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d49e812cfef9a381b16a22830542a40d3a42485b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-15 15:40:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6fdab9937543f79c175be0b06e690683f622fbf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 15:41:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7554881a824db3da24f0549a7394f9a56ef0bd41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 17:06:27.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/649727bcd9b94988a1eb659e580e24ce\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/052ec2be261b4178a498f45fdbd58657 : SUCCESS in 2m 50s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/561fc6eae67e48038ad0eea348666518 : SUCCESS in 2m 36s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/346c0a7cf25b428bad650ccc6d025085 : SUCCESS in 16m 18s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80caf04b59a9448e818bb77fbb765f99 : SUCCESS in 10m 20s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/f21ea921f5914e07a5ad55e61c43cd6e : SUCCESS in 16m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3a7dc6fe3d44820a4b1d1e7f1e25037 : SUCCESS in 58m 23s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78dfaf2000a04c8cac7e5e874cf254a2 : FAILURE in 1h 09m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/039fbe0b70e34dd59a80d35032aae4a5 : SUCCESS in 59m 47s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/b741205747c04951abc711ce5fb5bd0b : SUCCESS in 15m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4443007ca50d4e3dbe32f1c669cafc53 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"d6e353dabad57c2443961b32804da7a4e361076c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-15 17:09:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"695bbaf0352a1c48200d197bae1ad773ca8a7695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:23:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50d0b22e6b24f60abca61e92c93e7ba\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56d9b9384eae4dccacee25d096f43177 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b1a841d957441aa81e565002cbaefe1 : SUCCESS in 2m 53s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/2887fb0d3b0943be82735e60026afcca : SUCCESS in 6m 08s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/3deed860fb82457181d49f88e2330bb8 : SUCCESS in 15m 53s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba05af3f1f1d4fb7abde38ca1c78a9c2 : SUCCESS in 11m 57s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/83211f6872d04247a3680519636f395e : SUCCESS in 15m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/947564e780aa4111bd8a7ed1507c9b7a : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10928b447fde463b9e790fcb86606293 : SUCCESS in 45m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b61029c5eb9f4d73878be871a351dc50 : SUCCESS in 1h 05m 30s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b858244fae24ec89e5bba88dc11b26a : SUCCESS in 42m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/41dde5ec84a74985b91fca62ed0212b5 : SUCCESS in 1h 01m 15s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70ebdabdfa1741bcb4a2a65055f7e62d : SUCCESS in 50m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/53f8ea80a5d64ec0a9c92f55c0b0256a : SUCCESS in 53m 02s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/bdb3f53534ae47a7996e5c30dc17b0ac : SUCCESS in 14m 41s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0a51fccbaed492cb889625cbd228eb4 : SUCCESS in 9m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1ad27ec2b844ddea4a9201b8233b91e : SUCCESS in 9m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3018601daf84453287d11c67082921d8 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"d5d6428b4cdd6c42e22620f79727a3f1ce6e0bd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:24:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"04e72e32bab74351140180fa20f4a039f2e93336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 19:28:55.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c65e59f5b1546879afe524f2f3e06ea\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f37a41bf20144059bd7f82982990b30 : SUCCESS in 3m 02s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e7f5fa6fd6e4a5794e24d12cea2fa2b : SUCCESS in 2m 51s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/04b665df4ae24bf6bd299fc5325843d6 : SUCCESS in 15m 17s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/093afbe4d6f24c67b7985a1f1a01a324 : SUCCESS in 11m 43s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/30cf497098a94579bff04cdeeddac5c7 : SUCCESS in 13m 29s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c08f1053fa44ab1a0434dc9ffbfe16d : SUCCESS in 57m 06s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/414d7467149c491ca1ea59e1b6507ecf : SUCCESS in 43m 43s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/207fa440fdb747a29a796261a9c099f0 : FAILURE in 57m 33s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6a1c610371749d09c5d28633b6cdea7 : SUCCESS in 16m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8c654773ce6473ea779558c52a5b81c : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"3e53ac6cfff785db8f6f8ed9c73d406682f2b844","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-16 02:47:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4189ee3530a16797bc1647b7fdcf01e5bfda5cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 04:04:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd235a14ff84423c90faf0f65813321f\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da11e0f3cf194a0486e99bb70501f2c3 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd61ff5cd00450ba26c031d9d1b8b11 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/674e95eee4ba4b9c9d458e7d4ddadda1 : SUCCESS in 6m 00s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/edb91f517da842a595cd612f0e29f6e8 : SUCCESS in 16m 07s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20dd658a42104d15910ab3ac6c0a0f2f : SUCCESS in 12m 32s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/39d0f3d4bb504f4f982d547db7951604 : SUCCESS in 16m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bbc8bb1ae1a4dd18bfdbe40ab39c526 : SUCCESS in 12m 53s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da349edb98874f1f9305e4ce9b945fa7 : SUCCESS in 55m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9ea8f1025764babbc10eda5e22d4508 : SUCCESS in 1h 01m 10s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e36c86183f2345f4ab29627574c08bca : SUCCESS in 44m 38s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f49e5ed0a0994eeca06b02c24cab5cc4 : SUCCESS in 48m 28s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c05c6774f09427a90105e253dbb19de : SUCCESS in 46m 18s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/972ff27aa8bf47928ade5d0c54c45188 : SUCCESS in 49m 49s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bd31956a2ad4cb3b916209c99b0d29f : SUCCESS in 15m 08s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee17543a1f404df995050ca2156a2683 : SUCCESS in 10m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbd8338cefcf4c56a3d11c40ffcb405d : SUCCESS in 9m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b399c9cfd6f2449bbd128b6abfdd9a29 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"593e716e7a35f99a4cf57156e93153a304bea226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 04:05:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f3b6e798d2c42b0107cafeffba4f8dc2663585cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 05:11:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1269329362460b8864b04a88738fa8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c27e7788c48b4f7a9d22c7d1146993d6 : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30471313199f4180b77f0d090cb9b30e : SUCCESS in 2m 32s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/23fc0335b3d84b778824bea8c783c692 : SUCCESS in 15m 05s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9864d6d08e59485794b87674898f6842 : SUCCESS in 11m 00s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cd148145b6d40f382f5bb470ac4ac57 : SUCCESS in 15m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/232ff6c0def84013b97de697d81acb88 : SUCCESS in 1h 02m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7090f4df4d94ea3aadc5e609997ed2e : SUCCESS in 57m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d54004b09b7f48fdbfdd1443d89b5f43 : SUCCESS in 39m 45s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/c236c5bd049a4d81a222b5d78edc2e10 : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/914ad21589dc44ea84b4fafb7f4d8460 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"a43650d519f15258dd084a1d8160f8c63cdd0cc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 05:11:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4b8741f1dee66319910e351e615dfaccbb4fb639","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 05:12:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f517f0c3642493595a3b8292b08fb34\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/990e9ef68c3849fa83926b19e36d2951 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e1e446e689830a809cb3055bbbe8fb6a60ecd9d1","revisions":{"60def093c7c17f3fb0f4f4657fa0d115f70c3c22":{"kind":"REWORK","_number":1,"created":"2023-11-14 01:31:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/900839/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/39/900839/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/1"}}},"commit":{"parents":[{"commit":"d94647073a0766b0e41579c02c352efeac964709","subject":"Add TOTP support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/d94647073a0766b0e41579c02c352efeac964709"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 01:30:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 01:31:50.000000000","tz":540},"subject":"Support purging local_settings.d","message":"Support purging local_settings.d\n\nIn Debian, some files are installed into local_settings.d by their\npackages and these conflict with the options added to local_settings.\nThis introduces an option to purge the directory.\n\nChange-Id: I7e074f78b15ab22d86d6a6a10f573cfc4052023a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/60def093c7c17f3fb0f4f4657fa0d115f70c3c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/60def093c7c17f3fb0f4f4657fa0d115f70c3c22"}]},"branch":"refs/heads/master"},"f19b8009fc8c5e799b883f432cd22ea16a89d65e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-14 02:16:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/900839/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/39/900839/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/2"}}},"commit":{"parents":[{"commit":"2bc5b701e3822679c6f20c4adc80d653c5f03cac","subject":"Debian: Fix wrong wsgi script path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2bc5b701e3822679c6f20c4adc80d653c5f03cac"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 01:30:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 02:16:09.000000000","tz":0},"subject":"Support purging local_settings.d","message":"Support purging local_settings.d\n\nIn Debian, some files are installed into local_settings.d by their\npackages and these conflict with the options added to local_settings.\nThis introduces an option to purge the directory.\n\nChange-Id: I7e074f78b15ab22d86d6a6a10f573cfc4052023a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f19b8009fc8c5e799b883f432cd22ea16a89d65e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f19b8009fc8c5e799b883f432cd22ea16a89d65e"}]},"branch":"refs/heads/master","description":"Rebase"},"e1e446e689830a809cb3055bbbe8fb6a60ecd9d1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-14 08:25:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/900839/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/39/900839/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/39/900839/3"}}},"commit":{"parents":[{"commit":"cfe50237b6935c7db9aceca546432f62618caa81","subject":"Debian: Fix wrong wsgi script path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/cfe50237b6935c7db9aceca546432f62618caa81"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 01:30:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 08:25:48.000000000","tz":0},"subject":"Support purging local_settings.d","message":"Support purging local_settings.d\n\nIn Debian, some files are installed into local_settings.d by their\npackages and these conflict with the options added to local_settings.\nThis introduces an option to purge the directory.\n\nChange-Id: I7e074f78b15ab22d86d6a6a10f573cfc4052023a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e1e446e689830a809cb3055bbbe8fb6a60ecd9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/e1e446e689830a809cb3055bbbe8fb6a60ecd9d1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
