)]}'
{"id":"openstack%2Fpuppet-horizon~901355","triplet_id":"openstack%2Fpuppet-horizon~master~Ifc2b63e6e6e068daacccbe30478cccdd7a6447c3","project":"openstack/puppet-horizon","branch":"master","topic":"puppet-bookworm","hashtags":[],"change_id":"Ifc2b63e6e6e068daacccbe30478cccdd7a6447c3","subject":"Bump supported Debian version to 12 (Bookworm)","status":"MERGED","created":"2023-11-18 09:11:21.000000000","updated":"2023-11-18 09:35:32.000000000","submitted":"2023-11-18 09:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901355-puppet-bookworm","meta_rev_id":"4c847b4030feec412fe77d0e80b2c4ad6451c97e","_number":901355,"virtual_id_number":901355,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-18 09:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-11-18 09:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-11-18 09:15:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-18 09:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82b4173d165c89db55539da181c28851a1ae1b49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-18 09:11:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fc4fc9c9a014edbbed86d6dcf6dbf92391827d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-18 09:15:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bc6a9ee01ce85cf42598f4ea2f55e25c83c687ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-18 09:15:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"299e372bd70f6c6c14563e6ba9d21e187e3e18ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:25:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eadc85c76ee4755a7fe1e8b462c0d5e\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/12a647c48c064a0f9ae012679603f1ed : SUCCESS in 2m 19s\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b376dc633fdc4f25ab405b671075c334 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f4f7f3d0bea6d02cafee4dc701eed845ce6dc128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:26:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5acd57e21a5beedfdfb172f9efe1e56f19348886","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:35:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b173ee20b18146199664b175ec601136\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/66fed7ca584947e79894a5381cdd5869 : SUCCESS in 2m 27s\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e69c007f52e74c288a242dec92f3513b : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4c847b4030feec412fe77d0e80b2c4ad6451c97e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 09:35:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32c2099414e3dccaa748de215b420421450b15a8","revisions":{"7793739e13f87733984110916f2c9737692a47f3":{"kind":"REWORK","_number":1,"created":"2023-11-18 09:11:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/901355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/55/901355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/1"}}},"commit":{"parents":[{"commit":"2467a38063a5e4969583f502793b637f139c8132","subject":"Merge \"Create dedicated class for ironic-ui\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2467a38063a5e4969583f502793b637f139c8132"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-18 09:11:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-18 09:11:12.000000000","tz":540},"subject":"Bump supported Debian version to 12 (Bookworm)","message":"Bump supported Debian version to 12 (Bookworm)\n\nDebian 12 was released in June 2023. Recent versions of OpenStack\n(actually, 2023.1 and later) no longer support Debian 11.\n\nChange-Id: Ifc2b63e6e6e068daacccbe30478cccdd7a6447c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7793739e13f87733984110916f2c9737692a47f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/7793739e13f87733984110916f2c9737692a47f3"}]},"branch":"refs/heads/master"},"32c2099414e3dccaa748de215b420421450b15a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-18 09:15:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/901355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/55/901355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/55/901355/2"}}},"commit":{"parents":[{"commit":"035a17b444bf954a83c598263745206852a5c43e","subject":"Merge \"Deprecate support of MemcachedCache backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/035a17b444bf954a83c598263745206852a5c43e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-18 09:11:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-18 09:15:42.000000000","tz":0},"subject":"Bump supported Debian version to 12 (Bookworm)","message":"Bump supported Debian version to 12 (Bookworm)\n\nDebian 12 was released in June 2023. Recent versions of OpenStack\n(actually, 2023.1 and later) no longer support Debian 11.\n\nChange-Id: Ifc2b63e6e6e068daacccbe30478cccdd7a6447c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/32c2099414e3dccaa748de215b420421450b15a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/32c2099414e3dccaa748de215b420421450b15a8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
