)]}'
{"id":"openstack%2Fpuppet-horizon~917834","triplet_id":"openstack%2Fpuppet-horizon~master~I3d9b04c04828641535ff42866ee378b2b902ebe0","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-02 11:41:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d9b04c04828641535ff42866ee378b2b902ebe0","subject":"Add support for SIMULTANEOUS_SESSIONS options","status":"MERGED","created":"2024-05-01 08:26:26.000000000","updated":"2024-05-02 11:42:56.000000000","submitted":"2024-05-02 11:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917834","meta_rev_id":"eb800d3885445171b23b0733a5b84eb460ee6d85","_number":917834,"virtual_id_number":917834,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 11:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2024-05-01 15:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2024-05-02 10:16:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2024-05-02 10:16:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-01 09:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-01 15:39:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2024-05-02 10:16:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7201fa3e6a80351e8618519fd4f235c56cfa15cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 08:26:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"049d2eb4736c310260aff6e2ff532c11e23366d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 09:51:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef3375cc917b4ca7b45d466a567db058\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2f06ac50d7447c68a0ac46f90f9985e : FAILURE in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c334f34998a44f9b8c178dadf727bbf : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/270de918b6144b1b939e940f622a6142 : SUCCESS in 5m 22s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9f36032b77a47628075872dce6183c0 : FAILURE in 14m 00s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f084680e67684aa394998754d92d3848 : FAILURE in 11m 41s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/75468b952da348ba9875abe4d9313286 : FAILURE in 15m 47s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d7face18e5442da9db7493ae8ca822c : FAILURE in 12m 00s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c546ba52c1843fd90017a518f47f868 : SUCCESS in 1h 01m 03s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f402dd257a64b46998fb7a3e1ecf035 : SUCCESS in 1h 19m 04s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd616a3a95fa4230bb14e7c300f92818 : FAILURE in 48m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/03899391c43c4a35961759816a20ccbc : SUCCESS in 55m 21s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a56e0db809b24f399885d6520968eb9a : SUCCESS in 47m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a63c14fb22d34ccd9d20e9f74c8044f4 : SUCCESS in 56m 22s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d829d7d85104409b83c9aa1dddeabf4e : SUCCESS in 44m 44s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70ee90cd43474416b1fdbd818351cc65 : SUCCESS in 57m 59s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/88624f399fa1400f8c2ff723ae28b8cd : SUCCESS in 14m 53s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ed03b40c4544ddc9980e9783d6a2929 : SUCCESS in 9m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/018508cd62ec4909a7f879398440d361 : SUCCESS in 8m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47bfe93f9e5445d4ad72e4169d3acb40 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8e04bb1aee36cd2aba8e19ef5593c0aa6a2b69ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-01 11:50:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"190e734dc3c96276e99d76a3c5e7d94afa42232a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 13:16:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e043514d32534aaf8aff15956b5ddfa8\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a9b715c25944874be4c416bb5eedcca : SUCCESS in 2m 24s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c21c507ad44fc9a237cbfff161cf97 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/009f06f5a3cb44b58ef2061062fd9025 : SUCCESS in 5m 28s (non-voting)\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/4450bca779b9401fad5c8756e41018a0 : SUCCESS in 15m 14s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6548f93870934262b0a6e1b715403942 : SUCCESS in 8m 33s\n- puppet-openstack-unit-8.5-centos-9-stream https://zuul.opendev.org/t/openstack/build/c89504af952048f5bb2942f71bb9f718 : SUCCESS in 12m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24d8289ebd4041c5a17bf9ac18790d00 : SUCCESS in 11m 54s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3ec3be67a894ebe870e6a516f6192f2 : SUCCESS in 52m 45s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecec35ffa3d045729ea4bd0f7e1cf92f : SUCCESS in 1h 20m 48s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d12885b9954745f1916f74f07729141d : FAILURE in 48m 57s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00442564b0bd44a784ff3830e77fe4c4 : SUCCESS in 45m 14s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e481ab7b497f4a78893f63e3345695ec : SUCCESS in 45m 17s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a28f482ba5a74e8eaa44325d9714bf06 : SUCCESS in 55m 53s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1bfa5de8cf54390aae9e4457b8274a4 : SUCCESS in 40m 19s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21300bee01174407b6482e2c1c1e230c : SUCCESS in 53m 16s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3a5989dcae14c58adae5d07bf5fa2a4 : SUCCESS in 14m 40s\n- puppet-openstack-litmus-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d203dac5d72c4dbba4af761f79890ba5 : SUCCESS in 8m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c542883f69664bbb8a487fa076d32d4f : SUCCESS in 8m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c8206d177be43d4b016209e16794e1d : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ecb9ab9ffc25e5eb84748063a9255c63a7724380","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-05-01 15:39:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18b44fd8f5be06357441cfeb28f926e6cc7fd22a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-05-02 10:16:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4abd5afd8d4a4b02b0233991d27d4f78906e979","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:17:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0932fb0eaeb3903edcb15edefeda236ebbd52b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:41:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ce48b9ec544f0f9dd6918cf60dba8d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8321ec41c9804c86aeff21bb2fc73448 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d381f0a99d23476c88571b3f5f965000 : SUCCESS in 2m 56s\n- puppet-openstack-unit-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a12b95bf79342569eee5c6e1f12eab8 : SUCCESS in 13m 29s\n- puppet-openstack-unit-7.29-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81400b9989a74a0aa7a49f620e83bddc : SUCCESS in 13m 02s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f91008d21d045739176ce3c4e2e62aa : SUCCESS in 1h 08m 02s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5aaf299f04184034bf75bf613ae71382 : SUCCESS in 58m 40s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/16c24e1d19514cccae75203516cfc343 : SUCCESS in 49m 05s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a545132467fa4cc082b56783b35a1b7d : SUCCESS in 52m 16s\n- puppet-openstack-litmus-7.29-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfebf10f55944964ab8e29430b5633b7 : SUCCESS in 16m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dee51cb1b0134f2d8fa7be335c2d3e0c : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e65be99e91a2655e27ca01fa304aceeb874a190c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:41:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"eb800d3885445171b23b0733a5b84eb460ee6d85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:42:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21fd19c2516d4d569901b33415bf3582\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56929509edc74377b52cf883b906c178 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"505dc74dcf8bdfa7a3bfe079417bd1c1523d0b05","revisions":{"dd8b1de71f878ffef0f4895352419495bd32dd3b":{"kind":"REWORK","_number":1,"created":"2024-05-01 08:26:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/917834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/34/917834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/1"}}},"commit":{"parents":[{"commit":"1dcc7f7df5a07c63127fae82b78dd3ef5bc287fc","subject":"Merge \"Make sure python3-redis is installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1dcc7f7df5a07c63127fae82b78dd3ef5bc287fc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 08:25:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 08:25:42.000000000","tz":540},"subject":"Add support for SIMULTANEOUS_SESSIONS options","message":"Add support for SIMULTANEOUS_SESSIONS options\n\n... which was added in 21.1.0 (Yoga) release[1].\n\n[1] https://docs.openstack.org/horizon/latest/configuration/settings.html#simultaneous-sessions\n\nChange-Id: I3d9b04c04828641535ff42866ee378b2b902ebe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dd8b1de71f878ffef0f4895352419495bd32dd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dd8b1de71f878ffef0f4895352419495bd32dd3b"}]},"branch":"refs/heads/master"},"505dc74dcf8bdfa7a3bfe079417bd1c1523d0b05":{"kind":"REWORK","_number":2,"created":"2024-05-01 11:50:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/917834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/34/917834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/34/917834/2"}}},"commit":{"parents":[{"commit":"1dcc7f7df5a07c63127fae82b78dd3ef5bc287fc","subject":"Merge \"Make sure python3-redis is installed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/1dcc7f7df5a07c63127fae82b78dd3ef5bc287fc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 08:25:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-01 11:50:32.000000000","tz":540},"subject":"Add support for SIMULTANEOUS_SESSIONS options","message":"Add support for SIMULTANEOUS_SESSIONS options\n\n... which was added in 21.1.0 (Yoga) release[1].\n\n[1] https://docs.openstack.org/horizon/latest/configuration/settings.html#simultaneous-sessions\n\nChange-Id: I3d9b04c04828641535ff42866ee378b2b902ebe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/505dc74dcf8bdfa7a3bfe079417bd1c1523d0b05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/505dc74dcf8bdfa7a3bfe079417bd1c1523d0b05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
