)]}'
{"id":"openstack%2Fpuppet-horizon~931665","triplet_id":"openstack%2Fpuppet-horizon~master~I104225e0153ebe7fad611b463d48f904960a8a7a","project":"openstack/puppet-horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-29 18:00:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I104225e0153ebe7fad611b463d48f904960a8a7a","subject":"Accept an array for cache_tls_allowed_ciphers","status":"MERGED","created":"2024-10-07 15:07:55.000000000","updated":"2024-10-29 18:00:30.000000000","submitted":"2024-10-29 18:00:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931665","meta_rev_id":"529967b0cecbe8c8265dcc05ecd1a056ab3e3ba4","_number":931665,"virtual_id_number":931665,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-29 18:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-10-29 10:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-29 16:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-29 16:26:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-07 15:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-29 09:33:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-29 16:26:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f0a567acbc804385ddd6988839b46d730a261e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 15:07:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98158f0451a19227a7ddcdd525b3eb30724c7005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 15:24:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50d5511630964cb7be7ff73e7bb399e0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0021f67d8294b588b8099f9ae82d257 : SUCCESS in 2m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01fab02fec7e4cb38f3b47951fd4a5a6 : FAILURE in 3m 15s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdc2968512794bb8b3c91be7763d4698 : FAILURE in 5m 32s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed28a99640704426a55f88c5d6afea39 : FAILURE in 11m 11s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dab97fe2ce8444b9a61abbd446657ac : FAILURE in 7m 39s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/db851b8d03404708af28d6e71aba8144 : FAILURE in 10m 38s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c13ab7f2cc0b4a69af10cd2b5c7fcf12 : FAILURE in 8m 40s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01c9a7b066064248835626ab4c6cd759 : FAILURE in 5m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/923c74646c3e438eb6c5b13f2b879404 : FAILURE in 9m 06s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee6403811b314dba9a1a9859a0a30368 : FAILURE in 4m 50s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/01b3b0d65ceb43b7a3bb66028df48214 : FAILURE in 8m 07s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9886054908d4d58b67c2d1603ea9a2b : FAILURE in 4m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27b060a8bdf34794a824e30a800f0189 : FAILURE in 8m 26s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d14d5933b6ae450fa19a2e9b89458e76 : FAILURE in 4m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/620c2b614b604c22a3b93765a9619ef2 : FAILURE in 9m 14s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/fdba7641c9864280afbb3a72dcbc1922 : FAILURE in 6m 40s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e253c8b2e4434207b2450c8fae11a943 : FAILURE in 4m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f46dde23a2ce4dc6815747c588f34a34 : FAILURE in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba5912b24455d972d995e6e2931f417111c069c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-07 15:43:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7434e096b3c315a1a7167b0495223fb3baa0bda8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-07 16:00:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/feb81b28e1db4010ab1e813ce65b0cc9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2603cd5747c24389a43d7ef3c84a4ce8 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/509aababa88f4b1fb13ed9aff9dd4cca : FAILURE in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/f78cb4822cf8416192ef47111dff882d : FAILURE in 5m 25s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/3df7243820b5411b9f6402d0144b1671 : FAILURE in 7m 05s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75238f54b1f347dcbccd9db3f9e982f3 : FAILURE in 8m 19s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a929c618d1ec4754bf3753f874fd5ec8 : FAILURE in 12m 00s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17e4a33a7d2f4446b9b7916787ceb7e6 : FAILURE in 7m 52s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e406e127e7064ad99ba8345619e12b98 : FAILURE in 3m 47s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26c5072d58844b8a8b5ab3df4f31ef1d : FAILURE in 10m 23s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f96b1de1cda4591ad745940176156b9 : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/555a8537967d4cbaa4263245a0159eec : FAILURE in 9m 37s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3c5921c2d8944238237e91c633a51b7 : FAILURE in 4m 33s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/af2da120ae7c4df2befa218f20a3bbd5 : FAILURE in 9m 43s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b42d578a2a5439c90f7aca07c749221 : FAILURE in 5m 15s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e682a721db949339f07af2bf3140b20 : FAILURE in 9m 42s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f7f7388a62c43e791101217a746ac5b : FAILURE in 6m 55s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a7f88902a34481e8f400deb6878ec01 : FAILURE in 4m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af2d3000c3b44b9cb8d6998812d6518c : FAILURE in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57505826d214ea70b2603a1f6f84d548f0d6b2a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 05:36:02.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4f7b042ef5a0866daee3b40b1817d57be60d3c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 06:09:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38fdaabcff564baf84ad3e3e8dd8c9f9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6dde9e472757462e97bfff1268de98f0 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/187e427c44f043d6bccfd91d5a832096 : FAILURE in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3e330966a2e48cda9d8a3d48cb9aea4 : FAILURE in 5m 26s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6dfb35d556e4481b0b8a6df34b1587c : FAILURE in 8m 19s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aff9c868b66545d1bf92e100eddf3e11 : FAILURE in 6m 07s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/917d73df403345958c76702e5afcaf0f : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da041bcca0be4b398f8af898f95e059e : FAILURE in 6m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3f2950fc8d949d8824be1cecdf8365b : FAILURE in 4m 12s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2db4c5b3c0fb43308abe91e018e5591d : FAILURE in 10m 16s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a798a144d1d4fbebebecf3f2ca5ca03 : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c1e4cafa6a545ef8c3d22608d97763b : FAILURE in 6m 21s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c461d4ea58c54d31b8c6abdaefc6e8c0 : FAILURE in 5m 33s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cec85655279a42b3b88a662f4822530c : FAILURE in 10m 26s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40df8b71380343af8393794af0883833 : FAILURE in 5m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c74e0f826774636864cbd4c19dad7e6 : FAILURE in 11m 11s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/907391004fa04d858deefbab5a5d4dc2 : FAILURE in 7m 23s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85adb72763f047e7a079f0a882e1d9c4 : FAILURE in 4m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9de9e64cde0e4c6ab792d2c15aa08955 : FAILURE in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3deae94fe7ea60721e96c492c801db492f3a6bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-08 06:19:45.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b8de62438127fb9c26f7f4b95c37d78bd37d359d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 06:39:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2781e1fd5624401098a551ae5427285b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e26d3b23b944d79883d32802b74927c : SUCCESS in 2m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38b7eeec60a44137a6954880a2855ea1 : SUCCESS in 2m 10s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/df6e7d365340403aae3aa05c8e90d19d : SUCCESS in 3m 39s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b44e0aebf1ad4294bc39c55513b7b45e : FAILURE in 11m 31s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/003ab8d5559f40bcb895b800ef234357 : FAILURE in 6m 08s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3aa57516e3e649f688a63130d6bd6b51 : FAILURE in 9m 31s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51748966690d4eff81efa4f85e949c76 : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe69d4892a0b4af58aa57f0141339a3f : FAILURE in 6m 08s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e86899892ead411dac1741b45e6df543 : FAILURE in 10m 47s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f617aa52319d4d8a8f930122f10e2157 : FAILURE in 4m 58s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df6833dd8cb34edebb95407ea5f64d22 : FAILURE in 9m 21s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d90b15066e2b456fb3f1c8598050ec80 : FAILURE in 5m 04s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/869df678e4144ab9b62184f64c7b0bcf : FAILURE in 10m 19s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d341f819d824904b882fbe68551be63 : FAILURE in 5m 02s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4e6fd7af6074374ba309908ae6aeed8 : FAILURE in 10m 11s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5de54dca7c1845f0aec8567779d07ba7 : FAILURE in 7m 18s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad890376777c4cc79cc4e4800cc1b41f : FAILURE in 4m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a11c94baa74646b098c7ada793f8608c : FAILURE in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4dd3b009559fdfb0e41c118524f336c8f24ac730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 05:16:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e8e902de4c598a80c673e55a9c3ae342ee053320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 06:17:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8947530339b240478816c7bb883b3b12\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274801384e1c4c808f14057f00eb219b : SUCCESS in 2m 09s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1557bf76861b4ef687270a574d8024a4 : SUCCESS in 2m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/1440d93b244b47e6b833c519522c2305 : SUCCESS in 4m 22s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/570db451833b496b9f95fc9e8178653e : SUCCESS in 9m 46s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf0f8c3ee95f4ab1a0f8977d1d8502bf : SUCCESS in 6m 05s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e227df51868455db66881f57fbe329a : SUCCESS in 7m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0313e8f78f164abc9a6926e78584388f : SUCCESS in 6m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d8520100d524c4dbe2b712791191ce8 : SUCCESS in 36m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70b6962baffc4201908b96edd4052663 : SUCCESS in 44m 53s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fb3aa5af4cf4138a2033a6e48e323a6 : FAILURE in 30m 13s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/00f3d7d3ae4f460d9be49a18b91364f5 : SUCCESS in 50m 33s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6217836c4e64b86bdea2611c053fb7f : SUCCESS in 39m 18s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/132b7b75e0b04b9393022949d319f285 : SUCCESS in 39m 11s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a28291a56144fc1a71fc1b5f8734f5a : FAILURE in 25m 11s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30cdc6e4d72a4e70b6954f86e59494fc : SUCCESS in 55m 03s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbb9e5cdccf944aaaadcb5dbc24bc9b2 : SUCCESS in 13m 55s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d9d5f98d4cf4125bfd7977addc3812d : SUCCESS in 6m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b51a671c9ae42bab66d475a53c304e1 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"791632193917c6feabaa7c063cf7f2050cbb2356","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-29 09:33:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cf9884daa7040ec855ff2ac7212f7b77578e884d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-29 10:57:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4ce7a3da049627c8111b4a69a378797f799be82e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 12:13:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672e6c4706de45ed8efecd70f52a5538\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49c1b537a68a40f7a344eaab705b83e9 : SUCCESS in 2m 38s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c63fbe5d970401f807846b865b09459 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/994b4552a3f84be1abc306f0cebe61d8 : SUCCESS in 6m 16s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b086e07aceb45e897a557c29f3ad2d0 : SUCCESS in 14m 14s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d75a8846b7947e7a607602ac77a00ce : SUCCESS in 9m 58s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1fbcc39edb24721b92c4de3eb00450f : SUCCESS in 14m 13s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b13a7abb7284cdcbafa926ad79b0649 : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99d996ab445946f3bbfb8ec235e74f40 : SUCCESS in 56m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec40d2e3af074087ba2dd4f7f0d5e0ec : SUCCESS in 1h 12m 42s\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d51bb9b33444123a0313469576e48f4 : SUCCESS in 40m 19s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f23be143d76747f5847a8222b2908c44 : SUCCESS in 1h 02m 19s\n- puppet-openstack-integration-7-scenario004-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1975b92cd134845a5e45f6e2605d17a : SUCCESS in 42m 24s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/45138469d39d4ea5a3d01d27a331018b : SUCCESS in 54m 32s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27c5f7b4c1344746b3bf4f85dec614a0 : FAILURE in 1h 08m 32s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d96e62e70fb48dfb0670d06d920339c : SUCCESS in 41m 21s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f725f65e1a041f6baf1612374bad64a : SUCCESS in 8m 02s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5003af910fbc4f3b937074ac8cebdcaa : SUCCESS in 5m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b520df43c0d2494ba3221ee11ae0bd51 : SUCCESS in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7af9f4bafe5202237b8b76a109c0008e17e05e40","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-29 16:26:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9f1fc6aed613e74cf6a5e034c4bf72bbb724faae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 16:26:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8a858467d1224744a3b78c678f5a65bf66c52442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 18:00:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5b487c79c54421994cbcc6587e22452\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ca8179de43947ba89957f9f951f19ec : SUCCESS in 2m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09f007f2e9ae4a9ea20a215219017e07 : SUCCESS in 1m 53s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/abca891d2a0f450c8168043e2fb1adc7 : SUCCESS in 14m 29s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8562c15f2d4d1a883ec211cd6e4a5b : SUCCESS in 11m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/664d0d1e8bf54c438cb67ed415e5b995 : SUCCESS in 1h 28m 28s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f953f87aaef4550aa413a81fc62e872 : SUCCESS in 1h 06m 35s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/347d350066c049e5bb93b04f4bbc10f5 : SUCCESS in 1h 00m 26s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d67786a45b74133a3e0e4c66ca0ec4e : SUCCESS in 1h 02m 55s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/f802dbc4dd8f4e3691dee1c20347ef51 : SUCCESS in 14m 14s","accounts_in_message":[],"_revision_number":6},{"id":"529967b0cecbe8c8265dcc05ecd1a056ab3e3ba4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 18:00:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dfce4a9d13bec9bf9a3ec7ad9bba3bd0fe685c14","revisions":{"188dc2fb2999af315a02ebc03ad226f0ddc87418":{"kind":"REWORK","_number":1,"created":"2024-10-07 15:07:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/1"}}},"commit":{"parents":[{"commit":"507e4b29e8140442837b6f920017996a4794625a","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/507e4b29e8140442837b6f920017996a4794625a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:04:23.000000000","tz":540},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/188dc2fb2999af315a02ebc03ad226f0ddc87418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/188dc2fb2999af315a02ebc03ad226f0ddc87418"}]},"branch":"refs/heads/master"},"9db31512740307d7f76216e97b7cee8680d3410e":{"kind":"REWORK","_number":2,"created":"2024-10-07 15:43:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/2"}}},"commit":{"parents":[{"commit":"507e4b29e8140442837b6f920017996a4794625a","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/507e4b29e8140442837b6f920017996a4794625a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:33:25.000000000","tz":540},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9db31512740307d7f76216e97b7cee8680d3410e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/9db31512740307d7f76216e97b7cee8680d3410e"}]},"branch":"refs/heads/master"},"6b759dcfde8beaf158238d051b6558b403253856":{"kind":"REWORK","_number":3,"created":"2024-10-08 05:36:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/3"}}},"commit":{"parents":[{"commit":"507e4b29e8140442837b6f920017996a4794625a","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/507e4b29e8140442837b6f920017996a4794625a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 05:36:01.000000000","tz":0},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6b759dcfde8beaf158238d051b6558b403253856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/6b759dcfde8beaf158238d051b6558b403253856"}]},"branch":"refs/heads/master"},"2871a3154a97fc142253b52fbb061ded13594b14":{"kind":"REWORK","_number":4,"created":"2024-10-08 06:19:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/4"}}},"commit":{"parents":[{"commit":"507e4b29e8140442837b6f920017996a4794625a","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/507e4b29e8140442837b6f920017996a4794625a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-08 06:19:45.000000000","tz":0},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2871a3154a97fc142253b52fbb061ded13594b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/2871a3154a97fc142253b52fbb061ded13594b14"}]},"branch":"refs/heads/master"},"f2bb73ed3888c9a9dcbadfee261af5503b7f009c":{"kind":"REWORK","_number":5,"created":"2024-10-29 05:16:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/5"}}},"commit":{"parents":[{"commit":"507e4b29e8140442837b6f920017996a4794625a","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/507e4b29e8140442837b6f920017996a4794625a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-29 05:15:49.000000000","tz":540},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f2bb73ed3888c9a9dcbadfee261af5503b7f009c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/f2bb73ed3888c9a9dcbadfee261af5503b7f009c"}]},"branch":"refs/heads/master"},"dfce4a9d13bec9bf9a3ec7ad9bba3bd0fe685c14":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-29 10:57:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/931665/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-horizon","ref":"refs/changes/65/931665/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-horizon refs/changes/65/931665/6"}}},"commit":{"parents":[{"commit":"cd5216f5a59835485869f6d352e92c7473de3c5d","subject":"Pass custom_wsgi_process_options to horizon::wsgi::apache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/cd5216f5a59835485869f6d352e92c7473de3c5d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-07 15:03:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-29 10:55:57.000000000","tz":540},"subject":"Accept an array for cache_tls_allowed_ciphers","message":"Accept an array for cache_tls_allowed_ciphers\n\nThe parameter accepts openssl cipher list. Accept an array value and\nconvert it to the appropriate format to use the native type for easier\nhandling.\n\nChange-Id: I104225e0153ebe7fad611b463d48f904960a8a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dfce4a9d13bec9bf9a3ec7ad9bba3bd0fe685c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-horizon/commit/dfce4a9d13bec9bf9a3ec7ad9bba3bd0fe685c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
