)]}'
{"id":"openstack%2Fpuppet-ironic~659418","triplet_id":"openstack%2Fpuppet-ironic~master~I224d78a4de42a86df251ac7a0eca873f442f5ffc","project":"openstack/puppet-ironic","branch":"master","topic":"add_openstackclient","hashtags":[],"change_id":"I224d78a4de42a86df251ac7a0eca873f442f5ffc","subject":"Add openstackclient installation to the client class","status":"MERGED","created":"2019-05-16 01:53:39.000000000","updated":"2019-05-17 21:22:13.000000000","submitted":"2019-05-17 21:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659418-1558128133013-d1a6a7d1","meta_rev_id":"94c79243aa04d1f48f2df6bdf803265feab3ec1a","_number":659418,"virtual_id_number":659418,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-05-17 21:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-17 09:57:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2019-05-17 17:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2019-05-17 19:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-17 19:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-05-17 19:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 09:57:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-17 17:52:09.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2019-05-17 19:23:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-17 19:27:12.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-05-17 21:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e416c98959e652bba159d07b05fb3b0ed978b86","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-16 01:53:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635895bacf254d34e12d3bf882cbc025725febe9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 02:39:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/18/659418/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/0636724/ : SUCCESS in 37m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/18/659418/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/73482d4/ : SUCCESS in 41m 30s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/18/659418/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/d700ea4/ : SUCCESS in 41m 18s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/18/659418/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/57c990d/ : FAILURE in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1746556701f1949817bb57ceb91b545b6e1154ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 03:32:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-lint-ubuntu-bionic/b57f574/ : SUCCESS in 6m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-syntax-4-ubuntu-bionic/fc42dcf/ : SUCCESS in 6m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-syntax-5-ubuntu-bionic/b0cfbae/ : SUCCESS in 6m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-syntax-6-ubuntu-bionic/cef9547/ : SUCCESS in 6m 25s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/18/659418/1/check/puppet-openstack-unit-4.8-centos-7/5712c2e/ : SUCCESS in 14m 20s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/18/659418/1/check/puppet-openstack-unit-5.5-centos-7/bef9fb5/ : SUCCESS in 12m 47s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/f130ab8/ : SUCCESS in 11m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-unit-latest-ubuntu-bionic/48e53b2/ : SUCCESS in 12m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/fc089d5/ : FAILURE in 28m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/18/659418/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/a41bc95/ : SUCCESS in 1h 03m 52s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/18/659418/1/check/puppet-openstack-beaker-centos-7/c17e7b1/ : SUCCESS in 17m 22s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/18/659418/1/check/puppet-openstack-beaker-ubuntu-bionic/385bf43/ : FAILURE in 9m 44s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/18/659418/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/a6a0fd2/ : SUCCESS in 1h 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14ca76f319b2eb001490eb8a8c8bbf0a4406843c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 01:02:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"891ef0eb86b17c001c99775e0d09a99647ef811d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 02:42:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/18/659418/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/ab99b71/ : SUCCESS in 1h 31m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/18/659418/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/7ce63d7/ : SUCCESS in 1h 37m 17s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/18/659418/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/217c017/ : SUCCESS in 1h 36m 41s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/18/659418/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/2eef28c/ : SUCCESS in 1h 07m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5afc1620a351ca7ab6da028ddb529a8395efa4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 02:45:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-lint-ubuntu-bionic/4e8800d/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-syntax-4-ubuntu-bionic/c93cabf/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-syntax-5-ubuntu-bionic/8cab6b5/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-syntax-6-ubuntu-bionic/0bd050b/ : SUCCESS in 6m 25s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/18/659418/2/check/puppet-openstack-unit-4.8-centos-7/f922169/ : SUCCESS in 23m 16s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/18/659418/2/check/puppet-openstack-unit-5.5-centos-7/2e90e4e/ : SUCCESS in 20m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/bc4d2d3/ : SUCCESS in 9m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-unit-latest-ubuntu-bionic/04c32e6/ : SUCCESS in 11m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/39bde5b/ : FAILURE in 32m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/18/659418/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/12eb0c3/ : SUCCESS in 58m 44s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/18/659418/2/check/puppet-openstack-beaker-centos-7/d6a31e1/ : SUCCESS in 16m 05s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/18/659418/2/check/puppet-openstack-beaker-ubuntu-bionic/13de818/ : FAILURE in 10m 11s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/18/659418/2/check/tripleo-puppet-ci-centos-7-undercloud-containers/4274ddc/ : SUCCESS in 1h 28m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/18/659418/2/check/build-openstack-releasenotes/6012bf7/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"454c586c45e31f3154d353848b2a12de0829ddf7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-17 08:35:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e90815af8456e4a568e2e3a8e50effeccf800421","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 09:57:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/18/659418/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/f8533bc/ : SUCCESS in 1h 14m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/18/659418/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e3afa15/ : SUCCESS in 1h 19m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/18/659418/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/1fcb028/ : SUCCESS in 1h 19m 48s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/18/659418/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/2bd89bc/ : SUCCESS in 56m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd33b3eaefd8c21f25856583c588abd0e94c7de5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 10:13:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-lint-ubuntu-bionic/9640914/ : SUCCESS in 6m 50s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-syntax-4-ubuntu-bionic/77e376b/ : SUCCESS in 3m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-syntax-5-ubuntu-bionic/2190dd5/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-syntax-6-ubuntu-bionic/6370768/ : SUCCESS in 6m 22s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/18/659418/3/check/puppet-openstack-unit-4.8-centos-7/6eec751/ : SUCCESS in 20m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/18/659418/3/check/puppet-openstack-unit-5.5-centos-7/fe0ff67/ : SUCCESS in 21m 30s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/f563af7/ : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-unit-latest-ubuntu-bionic/843ff94/ : SUCCESS in 9m 19s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/8073436/ : FAILURE in 31m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/18/659418/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/7a6818a/ : SUCCESS in 57m 09s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/18/659418/3/check/puppet-openstack-beaker-centos-7/342bd40/ : SUCCESS in 15m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/18/659418/3/check/puppet-openstack-beaker-ubuntu-bionic/2288c45/ : FAILURE in 11m 45s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/18/659418/3/check/tripleo-puppet-ci-centos-7-undercloud-containers/6fd3769/ : SUCCESS in 1h 35m 38s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/18/659418/3/check/build-openstack-releasenotes/e24128a/html/ : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"0aa7aade455992b52fd2bb269b5488cdd0b4c38b","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-17 17:52:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7150305b4390e6260e3fea4816ca680940d7cb23","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-17 19:23:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18856e3e593284970f03f3e6193da7dfdcfe0906","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-17 19:27:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3a6ba931992f84d7691aebeb179829921d3606d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 19:27:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ebb31ef7b5c6ad129fb3e20374bdff39ef40ac0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:22:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-lint-ubuntu-bionic/5c3a4e4/ : SUCCESS in 6m 37s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-syntax-4-ubuntu-bionic/6e852f6/ : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/2466277/ : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/d794447/ : SUCCESS in 3m 48s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/18/659418/3/gate/puppet-openstack-unit-4.8-centos-7/0c4f9fb/ : SUCCESS in 21m 31s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/18/659418/3/gate/puppet-openstack-unit-5.5-centos-7/a0eb32b/ : SUCCESS in 20m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/7bf4e1a/ : SUCCESS in 12m 22s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/5e0e2ff/ : FAILURE in 32m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/18/659418/3/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/eda07ff/ : SUCCESS in 59m 06s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/18/659418/3/gate/puppet-openstack-beaker-centos-7/43ae12c/ : SUCCESS in 17m 57s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/18/659418/3/gate/puppet-openstack-beaker-ubuntu-bionic/67ff706/ : FAILURE in 8m 31s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/18/659418/3/gate/tripleo-puppet-ci-centos-7-undercloud-containers/6d3503c/ : SUCCESS in 1h 48m 05s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/18/659418/3/gate/build-openstack-releasenotes/c54f95e/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"b5ff104a007e445f697f6eef2354babfc6e12533","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 21:22:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"650a1e5465362f6d2df7c4ca1fe76a8c5cf6acc9","revisions":{"8657f618b3650222f1daf3a9fd943b3b99bcf58c":{"kind":"REWORK","_number":1,"created":"2019-05-16 01:53:39.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/18/659418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/18/659418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/1"}}},"commit":{"parents":[{"commit":"39115e63c25d82e2050a1c7bf787393d89a0393d","subject":"Remove ironic::inspector::debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/39115e63c25d82e2050a1c7bf787393d89a0393d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"subject":"Add openstackclient","message":"Add openstackclient\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I224d78a4de42a86df251ac7a0eca873f442f5ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8657f618b3650222f1daf3a9fd943b3b99bcf58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/8657f618b3650222f1daf3a9fd943b3b99bcf58c"}]},"branch":"refs/heads/master"},"9973ad0eb4d205927a00a7e3e5b9ae3c0dad6954":{"kind":"REWORK","_number":2,"created":"2019-05-17 01:02:31.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/18/659418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/18/659418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/2"}}},"commit":{"parents":[{"commit":"aadb6975a922330f102620e5bff39a2dc1093fad","subject":"Merge \"Add rcp_transport parameter for conductor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/aadb6975a922330f102620e5bff39a2dc1093fad"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 01:01:34.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I224d78a4de42a86df251ac7a0eca873f442f5ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9973ad0eb4d205927a00a7e3e5b9ae3c0dad6954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9973ad0eb4d205927a00a7e3e5b9ae3c0dad6954"}]},"branch":"refs/heads/master"},"650a1e5465362f6d2df7c4ca1fe76a8c5cf6acc9":{"kind":"REWORK","_number":3,"created":"2019-05-17 08:35:17.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/18/659418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/18/659418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/18/659418/3"}}},"commit":{"parents":[{"commit":"aadb6975a922330f102620e5bff39a2dc1093fad","subject":"Merge \"Add rcp_transport parameter for conductor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/aadb6975a922330f102620e5bff39a2dc1093fad"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-16 01:52:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-17 08:31:18.000000000","tz":480},"subject":"Add openstackclient installation to the client class","message":"Add openstackclient installation to the client class\n\nThe repo is inside the openstackclient plugin commands[1].\n\n[1]https://docs.openstack.org/python-openstackclient/latest/cli/plugin-commands.html\n\nChange-Id: I224d78a4de42a86df251ac7a0eca873f442f5ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/650a1e5465362f6d2df7c4ca1fe76a8c5cf6acc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/650a1e5465362f6d2df7c4ca1fe76a8c5cf6acc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
