)]}'
{"id":"openstack%2Fpuppet-ironic~677632","triplet_id":"openstack%2Fpuppet-ironic~master~I768827ca462e25fcb0571a3b5cb5e5eca615cc08","project":"openstack/puppet-ironic","branch":"master","topic":"bug/1840868","hashtags":[],"change_id":"I768827ca462e25fcb0571a3b5cb5e5eca615cc08","subject":"Introduce the new rabbit_heartbeat_in_pthread option","status":"MERGED","created":"2019-08-21 06:25:19.000000000","updated":"2019-08-22 04:38:11.000000000","submitted":"2019-08-22 04:38:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677632-1566448691717-e43a0d63","meta_rev_id":"96ba5b0f24d91e3839aac86195a32a26a68d4ef1","_number":677632,"virtual_id_number":677632,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-08-22 04:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-21 08:24:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 02:24:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-21 15:17:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-22 02:24:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 08:24:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-21 15:17:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-08-22 02:24:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 04:38:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f22b68104c64c037fac47d86853c8fb2342b6ab4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-21 06:25:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92255ef9d3220e69175e7695a5f24d3d8042e3e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb9697e3160842da88c9f171edf242ec : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa164415f5734ed4bf579984e653e175 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94f9ef0ae7da4881acf0cc4d35c02ab0 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ed2d63a76e3242988e78e29b021ef820 : SUCCESS in 24m 20s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/925a7da13e05415cadf96d5302475bbc : SUCCESS in 16m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1589aab1927473092a64143cc9801d3 : SUCCESS in 17m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85a619eef25a432d885f597d34466aac : FAILURE in 36m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da045ab9adb14c57a8843753a1f92781 : SUCCESS in 1h 04m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4d4760a3824a45ada90e25cf02bb72c1 : SUCCESS in 20m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37479b07e5524b89acc97cf60d4cef40 : FAILURE in 8m 07s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/9466f9acb8c3471cae2886f44ad4d4a0 : FAILURE in 9m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5d706b81ba3466083883e69c067ea68 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"aa48165f70733178fdfbe7c7539dd5fc782a20a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 08:24:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/32/677632/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/d311b34/ : SUCCESS in 46m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/32/677632/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/f05d9b2/ : SUCCESS in 53m 02s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/32/677632/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3e58067/ : SUCCESS in 1h 12m 44s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/32/677632/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/a914414/ : SUCCESS in 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a3c4e85d6dcc0b8a6c9169f9622571a07178555","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-21 15:17:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fdc20e2da394f800c006da83dc6563d00b461da3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 02:24:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bff0db6f782ad2a7ef40aa8c38eea637300197a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 02:24:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"834a0c0587615f53daca847669cef2cf73bd772d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 04:38:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84e60e0484ba4b9b9e2eb79da24c46c6 : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce8ad0050180446d80cb9e6f2b4bedc4 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e875dfba4f184d339877f54b799ebf06 : SUCCESS in 6m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2781f32f4cd14626b8ca0a6891e2fbe7 : SUCCESS in 17m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a0cda1d7da94edabe08bb9905285df7 : SUCCESS in 10m 20s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9219701d36942b690cb03cfdf9eb98b : FAILURE in 34m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f9daa3f2d44e48c8a5072424be619cbb : SUCCESS in 57m 55s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/788ab8472e834697b4d0b86b4f06f313 : SUCCESS in 15m 54s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1620058d9bdc4693ac67a02302ecacaa : SUCCESS in 1h 44m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3f8124dff044830a960a3569aa6a3a9 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f01bd94157b1a718ded63e8b42cb28225db44e0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 04:38:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"486e62796ec2d6d2dcc2221f3e37ae5bfdb41549","revisions":{"486e62796ec2d6d2dcc2221f3e37ae5bfdb41549":{"kind":"REWORK","_number":1,"created":"2019-08-21 06:25:19.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/32/677632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/32/677632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/32/677632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/32/677632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/32/677632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/32/677632/1"}}},"commit":{"parents":[{"commit":"3c1fcfd2dcac12ccc6312d6e61f7c13c6e9a878e","subject":"Remove deprecated pki related options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3c1fcfd2dcac12ccc6312d6e61f7c13c6e9a878e"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:21:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-21 06:23:42.000000000","tz":480},"subject":"Introduce the new rabbit_heartbeat_in_pthread option","message":"Introduce the new rabbit_heartbeat_in_pthread option\n\noslo.messaging RabbitMQ driver have now a new option that allow user to\nrun the RabbitMQ heartbeat over a native python thread.\n\nThese change allow user to use this new option.\n\nChange-Id: I768827ca462e25fcb0571a3b5cb5e5eca615cc08\nCloses-Bug: #1840868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/486e62796ec2d6d2dcc2221f3e37ae5bfdb41549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/486e62796ec2d6d2dcc2221f3e37ae5bfdb41549"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
