)]}'
{"id":"openstack%2Fpuppet-ironic~685335","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fstein~Ic95dbf1867fb5397f1b6d8f23466910a97051cb6","project":"openstack/puppet-ironic","branch":"stable/stein","topic":"bug/1845487","hashtags":[],"change_id":"Ic95dbf1867fb5397f1b6d8f23466910a97051cb6","subject":"Add support to set option:mtu in inspector dnsmasq","status":"MERGED","created":"2019-09-27 12:16:21.000000000","updated":"2019-10-08 00:12:21.000000000","submitted":"2019-10-08 00:11:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685335-1570493469166-6aa585c6","meta_rev_id":"7d02663ed515703bdc093e2dd895b3539f2baae2","_number":685335,"virtual_id_number":685335,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"date":"2019-10-01 11:40:54.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 00:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-10-04 11:18:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2019-10-03 14:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-07 15:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 14:11:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-04 11:18:08.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-10-07 15:25:37.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2019-10-08 00:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41867af72ef70aa2ccab43b7b24621382f868239","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-27 12:16:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b6955308f489a34d11e5efd30cdd9f10dfc37988","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-27 12:26:06.000000000","message":"Topic changed from bug/1845487-stable/stein to bug/1845487","accounts_in_message":[],"_revision_number":1},{"id":"b4bb36ceca94f4872a48fedf4139a9f9f4112ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 14:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af6920b636524139b9f71d7bd50c6644 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee40c0977df14feda49e7e0915430ddc : SUCCESS in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05c23dbddb284996bcec2a83aa37fd1c : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6056e96191d548bcb08aead23c93cd24 : SUCCESS in 5m 09s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/7f4c0c8ff01a4df186e4b65be90c3be8 : SUCCESS in 23m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b1571958e8b84f7bbc48970917c7e3ab : SUCCESS in 23m 28s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47e600990fec40b18853369c3f7bec2b : FAILURE in 22m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1c36658052474b0d9cbae66d1537affe : SUCCESS in 1h 06m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a6d211b8c23046b3a822cd38dd5c7778 : FAILURE in 6m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ad815eba5054412a7887ed88118c0e3 : FAILURE in 9m 29s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca6930b898954fc98b538a0c47c607be : SUCCESS in 1h 54m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcbf629c30a6428db1912189b3aa7846 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"253050be70546e029506e7636bf71c80edb078b5","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-10-01 11:40:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c41f9d1d7b8fc7fb0aa365485b1ed5170c1a0332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 13:53:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9181eca33f4044e18b5be5f247081436 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c983527167d4137a2cdf66bbd304c3d : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4dab08fb9d7465ebebc8c9859250937 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c094971029f74d61a1d39e3b200b1658 : SUCCESS in 3m 32s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/ede95675f2d44c02954cd7ba8fe65372 : SUCCESS in 21m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b7087672c1e049cea1c9c8a30c1e7a49 : SUCCESS in 21m 14s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05a241a1f92f42678e253f97549b50dc : FAILURE in 21m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1917d041e746445b9e3344463d441a29 : SUCCESS in 1h 12m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ff1208f40c9c4df48091aa7cfc73409c : SUCCESS in 16m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b62d1beec1f44d0793bc3fd622ed40bf : FAILURE in 8m 57s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/28e16bbf6b424a4aa8ecbaca9bbe9042 : SUCCESS in 2h 04m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9689efe6c389437e8a71c23705ba7707 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a4a9e2cfd5089ec8c6b2c70648736e7235ef18fc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-03 14:11:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7e0a5510378471c7d183e56adb0664bb3eb995ce","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-04 11:18:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9e40e6526662eb1fab80ccd3580a451258d336e4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-07 15:25:37.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c6d23ceec3f6485e4b7ac373b4d2cfa4f278c5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 15:25:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58a75c04349f71ddfeedf1478d45c0883ef7775f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 16:41:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51a01febf60248bda10ec612ee94e1af : SUCCESS in 4m 30s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6956c6d9d1dc4184a859313123111ca4 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94685670bd0e4a95b0a41bacba1f53ec : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d57267c6fcbe46c6820df66a3c8b96d6 : SUCCESS in 4m 32s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/64e84bd9b58b4a0789493dc9abdfecf8 : FAILURE in 11m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7ad40db9e99e4e0fad03123021808f3e : FAILURE in 9m 12s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc6b62bcc56f485dbfc7fe6ad0c71da9 : FAILURE in 23m 13s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f6245f2ba1b4750819846cadd8656ba : SUCCESS in 1h 08m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8afa1b785ee94b72ab179232b24c0ada : SUCCESS in 18m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24ff5c10727a40a380cf8e457311b159 : FAILURE in 15m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e21dd391b5a4e58aa7697fb894c28ad : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"42d3ad22b5517324a86ef0d0a1db9836e5c9977f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:00:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5737f5aaf6c3cef8ac350222fd1d64f75556a33d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b280a340f26245b8b2250e50331e7450 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93c75fd406564b19945b5b9b2f37ca12 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/440018a8cd574b5bbf7f009f79b51ec9 : SUCCESS in 6m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5711e6469354c918fd48d321503f905 : SUCCESS in 6m 11s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1eda22d52f6441c09dbcc4bfc51e23a6 : SUCCESS in 18m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5696378688df477e91263469e51a5fa3 : SUCCESS in 18m 10s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/100510985f7148c0860e2b20b343854c : FAILURE in 27m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1f03eef44ba84870b7641f8a71ca030f : SUCCESS in 59m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aff0f4b75a0746df8123645ffb654068 : SUCCESS in 18m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/375c65681bdc462e8034a28777e21865 : FAILURE in 10m 49s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7e1bcce7a3e44b6fbfabb333132988b5 : SUCCESS in 1h 54m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436bbe4f192d4e63897f9905b5a15e5a : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e2248b2a3fd0bb0ec0e517b747801328ef9385d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:57:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42b3231385b95a5b103cd9cae5b32060e1298235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:11:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9fc82fc2bbc4c718fc490a92b3b7118 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78c4d769050c41689de177d912a01a79 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/381539f1ac5c458ca146e59f45662cfb : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28441fe98134484082bf62b077388e67 : SUCCESS in 4m 11s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/41262504329a4b2daa6160958344a13e : SUCCESS in 16m 52s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9b9c50aa57324c1f9220d5517d511027 : SUCCESS in 21m 11s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/632338fc9d804ccc97d6ca988c243d90 : FAILURE in 19m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da9461f3b240494aa2391f3c18f60169 : SUCCESS in 1h 03m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/deedf07709f24dd78121effa7aed24dd : SUCCESS in 18m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9f1ffd723224382a699bd295816dc50 : FAILURE in 7m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b276bd5839f4139bd23ce9bc311328b : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fd488258b3879af83d8bba737a51a686bccbc10a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:11:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d02663ed515703bdc093e2dd895b3539f2baae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:12:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8d33c5d1c36423ca1e5196900a968af : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"987a46a15dc1cfa541b73c7093ce362569d049dd","revisions":{"987a46a15dc1cfa541b73c7093ce362569d049dd":{"kind":"REWORK","_number":1,"created":"2019-09-27 12:16:21.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/35/685335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/35/685335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/35/685335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/35/685335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/35/685335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/35/685335/1"}}},"commit":{"parents":[{"commit":"cd9417ed8b405f1c4e9af34530410ffca59789dc","subject":"Also copy ldlinux.c32","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/cd9417ed8b405f1c4e9af34530410ffca59789dc"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-26 11:40:08.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-27 12:16:21.000000000","tz":0},"subject":"Add support to set option:mtu in inspector dnsmasq","message":"Add support to set option:mtu in inspector dnsmasq\n\nIt should be possible to configure the DHCP server to\nprovide the mtu option for inspection subnets.\n\nThis uses the dhcp-option-force so that the option is\nprovided also when the client does not ask for it.\nAccording to dnsmasq manual page this is sometimes\nrequired when sending options to PXELINUX.\n\nCloses-Bug: #1845487\nChange-Id: Ic95dbf1867fb5397f1b6d8f23466910a97051cb6\n(cherry picked from commit 03550bc8c8d05583e76c335c465684cec6d55093)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/987a46a15dc1cfa541b73c7093ce362569d049dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/987a46a15dc1cfa541b73c7093ce362569d049dd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
