)]}'
{"id":"openstack%2Fpuppet-ironic~685341","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fqueens~Ic95dbf1867fb5397f1b6d8f23466910a97051cb6","project":"openstack/puppet-ironic","branch":"stable/queens","topic":"bug/1845487","hashtags":[],"change_id":"Ic95dbf1867fb5397f1b6d8f23466910a97051cb6","subject":"Add support to set option:mtu in inspector dnsmasq","status":"MERGED","created":"2019-09-27 12:25:45.000000000","updated":"2019-10-08 00:12:20.000000000","submitted":"2019-10-08 00:11:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685341-1570493468558-d762e54f","meta_rev_id":"15c487ea943dd1fe25c25c66f165897e6d6a0d7e","_number":685341,"virtual_id_number":685341,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-10-07 21:01:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 00:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"recommended":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2019-10-04 11:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2019-09-30 19:26:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-07 15:25:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-30 19:26:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-04 11:18:27.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2019-10-07 15:25:56.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2019-10-07 21:01:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-08 00:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8547dd5feaa3d26f1db1ca9abfaf9e311c592420","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-27 12:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be5795e5adedb1ced807a3fe7e77fc75a150012c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 14:03:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/7fcada06ec404e1aa989e51bac32589a : SUCCESS in 6m 10s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/befaa958a8db4452afffda4b51a3f321 : SUCCESS in 5m 26s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/01d2cb687bf84610b20e51f6e384cfc4 : FAILURE in 7m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0fc0aa3c3dd54b629d99e082398f3931 : FAILURE in 16m 46s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a1b420f6a9794230bb796f4037203dc5 : TIMED_OUT in 1h 31m 16s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3fc0ce9c0a5341b6943b72bb4e7900d1 : SUCCESS in 57m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0f3c706a3e524b018df10c2d6313a4d3 : SUCCESS in 22m 08s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/844c1b46948f4fb7bb87aa57d364604f : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69bc420a834445abb9b08cd307275f5c : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f975aa0204b836c8a280287a4f159e1cbc2e3464","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-09-27 14:52:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5dc06b17a2c7090f08c13f209c7051b837abd90e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 16:29:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/4c7a209039534237b4a7df9ba49742e0 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/caeeeec08ad04b96bf51dc5a53760e59 : SUCCESS in 4m 14s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/7bce140f041c4326b711e38ff5c1dc67 : SUCCESS in 22m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6b29485b7fd0487dbe3bf51963211257 : SUCCESS in 15m 35s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f59335965c104fc08f5ee59b6187a511 : TIMED_OUT in 1h 30m 58s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/80f2350c34b54333961c44e842b85822 : SUCCESS in 1h 00m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9e3bf15f6b2b472081d600e038532dac : SUCCESS in 18m 34s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/a1dc019a60ea48dd82c0213388d0f7ec : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6522e50c7bf24590bce4e2c7b351e566 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"76db2015ec2db37a16a7708c9011fc82e7adc15b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-30 19:26:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a01412a6079c36dfd0fcda863f77f80d462f6985","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2019-10-04 11:18:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73e9a0ef3c1780de8362f7540757b4938ca1b894","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2019-10-07 15:25:56.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54280970726a3024e5e26d6fcbcb8a16d1a650eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 15:26:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef36a21c225b79fede23d1dbf13d837364336c09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 17:24:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/1caf331230324c33b572be5f2c4cbe38 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/fdc0c3554e4b4efb972495be84e19ce7 : SUCCESS in 4m 15s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/8e5f926371dd45e99359d59ad3f479ec : SUCCESS in 22m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ae01768ffae6424fbe7211ec66a928ba : SUCCESS in 22m 29s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/22f4554d90e14744b670d12724746e96 : SUCCESS in 1h 15m 01s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/e704f1897a45481f8c05029cb4fc0205 : TIMED_OUT in 1h 30m 55s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e066c1ec38c7488a83b99424de926628 : FAILURE in 5m 38s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/c81dd442238a4a338613b016200f7c0d : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7dab735c3794dd7b42d7b74ef8cb81e : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":2},{"id":"20fa6cdc4ad317e13f9e9076760b7b35689978aa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:01:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9be11a4218ba5ba63b55001368d806841e63ebcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:38:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/0a75b620336440bba31a0f97a9a0fb2d : SUCCESS in 3m 33s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/40d9ee3ed0ca465a83895ea69a7166b5 : SUCCESS in 3m 59s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/9291fe7905f447c9b7aabcbb264fc2df : SUCCESS in 13m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/01bd441fee134654b150ce59db697ceb : SUCCESS in 23m 54s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a62f5ee919fa4377ba4ec2adebb743ee : TIMED_OUT in 1h 31m 07s (non-voting)\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6e84762e613043d6ab3e2e46f8da1247 : SUCCESS in 58m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/34ca7d4856c9479f8d660b8484e168f1 : SUCCESS in 17m 51s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/71958b9ef6624cbc8d01185bee4e5f0d : SUCCESS in 15m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9208322f97044a749ab0f90e905e5caf : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1a4a2893107f652ab75151ac3abdd995057306dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:38:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08bd43ad05b4e81fd1908dc58cecdabe95af319f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:11:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/6a98b0984c114cf19b3ab5b4008994e5 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/20a26916eacd4853ae17bf769f9f4b42 : SUCCESS in 3m 46s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/707e3c1b4cf84599a7eb89599f8d90b5 : SUCCESS in 21m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1c8e2be87d954207904d4fb45d170503 : SUCCESS in 20m 05s\n- puppet-openstack-integration-4-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/75a27c2f38c4423aa8967445bdfc1963 : SUCCESS in 57m 53s\n- puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d942426dbebe4a0699e92d61de09f6d7 : TIMED_OUT in 1h 31m 09s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3625dba14888482e8cc6a36d7eecfddf : SUCCESS in 17m 08s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/28966c2ac1fd4fb08e3278e437ec26c6 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01eeb0c76a18459aaccb3ca7d07756e0 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"dd872794e459e84d82c84724f390cfc739da07b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:11:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a7b8fb57a8f5d1e85334b6c6d86c48d84f8add10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 00:12:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34730345211d4fac95a33abb746f0b59 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62286d200a8a15003165aea69515d2441aeddcd8","revisions":{"837190887c1f5fad0175f73708859cab0bb9667e":{"kind":"REWORK","_number":1,"created":"2019-09-27 12:25:45.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/685341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/41/685341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/1"}}},"commit":{"parents":[{"commit":"b870936fdf15796ba2b52b9aef8d95770b0dccf2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b870936fdf15796ba2b52b9aef8d95770b0dccf2"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-26 11:40:08.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-27 12:25:33.000000000","tz":120},"subject":"Add support to set option:mtu in inspector dnsmasq","message":"Add support to set option:mtu in inspector dnsmasq\n\nIt should be possible to configure the DHCP server to\nprovide the mtu option for inspection subnets.\n\nThis uses the dhcp-option-force so that the option is\nprovided also when the client does not ask for it.\nAccording to dnsmasq manual page this is sometimes\nrequired when sending options to PXELINUX.\n\nCloses-Bug: #1845487\nChange-Id: Ic95dbf1867fb5397f1b6d8f23466910a97051cb6\n(cherry picked from commit 03550bc8c8d05583e76c335c465684cec6d55093)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/837190887c1f5fad0175f73708859cab0bb9667e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/837190887c1f5fad0175f73708859cab0bb9667e"}]},"branch":"refs/heads/stable/queens"},"62286d200a8a15003165aea69515d2441aeddcd8":{"kind":"REWORK","_number":2,"created":"2019-09-27 14:52:27.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/685341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/41/685341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/41/685341/2"}}},"commit":{"parents":[{"commit":"b870936fdf15796ba2b52b9aef8d95770b0dccf2","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b870936fdf15796ba2b52b9aef8d95770b0dccf2"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-26 11:40:08.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-09-27 14:52:13.000000000","tz":120},"subject":"Add support to set option:mtu in inspector dnsmasq","message":"Add support to set option:mtu in inspector dnsmasq\n\nIt should be possible to configure the DHCP server to\nprovide the mtu option for inspection subnets.\n\nThis uses the dhcp-option-force so that the option is\nprovided also when the client does not ask for it.\nAccording to dnsmasq manual page this is sometimes\nrequired when sending options to PXELINUX.\n\nCloses-Bug: #1845487\nChange-Id: Ic95dbf1867fb5397f1b6d8f23466910a97051cb6\n(cherry picked from commit 03550bc8c8d05583e76c335c465684cec6d55093)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/62286d200a8a15003165aea69515d2441aeddcd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/62286d200a8a15003165aea69515d2441aeddcd8"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
