)]}'
{"id":"openstack%2Fpuppet-ironic~692995","triplet_id":"openstack%2Fpuppet-ironic~master~I4ddbba19f48d7f6643dab71097d2018b4a9757d3","project":"openstack/puppet-ironic","branch":"master","topic":"story/1528920","hashtags":[],"change_id":"I4ddbba19f48d7f6643dab71097d2018b4a9757d3","subject":"Configure parameters for managed in-band inspection","status":"MERGED","created":"2019-11-05 12:21:42.000000000","updated":"2019-11-26 15:46:12.000000000","submitted":"2019-11-26 15:44:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692995-1574783041407-a496186d","meta_rev_id":"0a46ea9c0b5a66816c1b764ca8de67bcbb358498","_number":692995,"virtual_id_number":692995,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 15:44:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-05 13:17:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-11-06 10:21:02.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-11-08 01:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-11-05 17:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-11-08 01:30:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-26 13:52:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 13:17:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-05 17:00:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-06 10:21:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-11-08 01:30:37.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-11-26 15:44:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5c04ba04255b49e60a39ce344e29630bb2917bc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-05 12:21:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bf6751e31ab5c29d9fd2da60efdb8d212fde910","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-05 13:17:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/95/692995/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/7a19e6c/ : SUCCESS in 36m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/95/692995/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/4817f64/ : SUCCESS in 54m 24s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/95/692995/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/6ed2833/ : FAILURE in 48m 35s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/95/692995/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/4826492/ : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2245c1cd4a96978c938345f1176956b99d2d8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 14:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3836938b55344c98ba1ea8daabef4add : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/696e32537c25426eac178951f3f9fc68 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6cebca025f549369a839ac46f17c029 : SUCCESS in 4m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8b7f4bdf4a7746b0be1acc80e04383c5 : SUCCESS in 20m 48s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb6616cbcd6f42239121d90b9b6f388e : SUCCESS in 13m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/788851c5f5be47fcafb76a06c9dd9330 : SUCCESS in 11m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfd0a5f6a11049a0973c41ab944ceb54 : SUCCESS in 44m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cbf8b2198bfa442491de5c0156db7cf2 : SUCCESS in 59m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6df9c19d352c4474b675233a9a56174e : SUCCESS in 18m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54a4debef8924535824d1d613ecfe8b3 : FAILURE in 16m 27s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/26d08e2c92d64a84821d709734ec6e07 : SUCCESS in 1h 43m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1248b995bf484e83964980e38d61a0 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6e6d254ddad99eeb3cf11c914d665815bb7fd1f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-05 17:00:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4de6b49e7819ed583394b1bddec5d220219ea9a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-11-06 10:21:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0897cf348ac1c8e991a832191a67e711c422822","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-11-06 10:21:38.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"435a37d0bec5d92b4ff3e67951e82ee3e5f22d52","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-11-08 01:30:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c245431e0b31d51ca74e19573a9fa38ca25df8cd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-11-26 13:52:34.000000000","message":"Patch Set 1: Workflow+1\n\nKicking zuul now that the dependency has merged.","accounts_in_message":[],"_revision_number":1},{"id":"86c48ed870b93b49af427159778217fd801375ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:52:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a922a56fa3fbf4cff704c31d4334fec3734ce844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:44:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63bb76d83b2f1f0dfe2fa9df449d6355d7b626c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:44:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6a66a5bb68407d9f1dadbdcdd2502f : SUCCESS in 8m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca2e6c4e01da4564bf4980d11839c54b : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec760f6c92b5483dbb9a064718a0260d : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e3567c6b6feb4d1ea3e0def416f52c69 : SUCCESS in 20m 15s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d0b319f41514f1aba30567cf2764577 : SUCCESS in 15m 51s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4925966d84af428f8c3d6606c5cff798 : SUCCESS in 46m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b769a25a07b44fd6afec28f9519e6dd2 : SUCCESS in 1h 00m 41s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/db748d10a4364c39b5cdd70116293ae2 : SUCCESS in 19m 51s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/59cf52ecbc1a4739a4d78f032a2c13fb : SUCCESS in 1h 49m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7b761a831594b5bb27b9ac98a2a4909 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"89bbf6b45737c70da419e223cd490218a7909941","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:46:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806b15c84eb746509c6f05a69e4f76d2 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7f0988c0c019d1ab09bb3d0c9343b96574069fb","revisions":{"b7f0988c0c019d1ab09bb3d0c9343b96574069fb":{"kind":"REWORK","_number":1,"created":"2019-11-05 12:21:42.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/95/692995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/95/692995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/95/692995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/95/692995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/95/692995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/95/692995/1"}}},"commit":{"parents":[{"commit":"6deffc3188222be306c21b845f994e2f701a0cd0","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/6deffc3188222be306c21b845f994e2f701a0cd0"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-11-05 12:17:13.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-11-05 12:21:25.000000000","tz":60},"subject":"Configure parameters for managed in-band inspection","message":"Configure parameters for managed in-band inspection\n\nChange-Id: I4ddbba19f48d7f6643dab71097d2018b4a9757d3\nDepends-On: https://review.opendev.org/688720\nStory: #1528920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b7f0988c0c019d1ab09bb3d0c9343b96574069fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b7f0988c0c019d1ab09bb3d0c9343b96574069fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
