)]}'
{"id":"openstack%2Fpuppet-ironic~720441","triplet_id":"openstack%2Fpuppet-ironic~stable%2Ftrain~I09e6ce611b12e92a4a8db64c7f23a491998e956b","project":"openstack/puppet-ironic","branch":"stable/train","topic":"bug/1872992","hashtags":[],"change_id":"I09e6ce611b12e92a4a8db64c7f23a491998e956b","subject":"Add ironic::neutron::dhcpv6_stateful_address_count","status":"MERGED","created":"2020-04-16 11:32:22.000000000","updated":"2020-04-17 10:40:09.000000000","submitted":"2020-04-17 10:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720441-1587119793192-767da42d","meta_rev_id":"e0fa8210c4b2395bf9ebab658a5eac01cbee7906","_number":720441,"virtual_id_number":720441,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 10:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-17 06:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-04-17 08:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2020-04-17 08:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 06:24:59.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-04-17 08:37:26.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-17 10:36:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d31b2a5524a286e4caecea69aae46e85fd4449b","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-16 11:32:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2e8e40b7f2e21dfa123752c66d9439fed506951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 13:40:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efea3685a6864017937e233fa51314fe : FAILURE in 4m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7719bc06b75f4372a839d1df45c9ce76 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93b12a29f71942368423ac3367ae9cfa : SUCCESS in 5m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/87bbaafb38ff42129935a07ffe33344c : SUCCESS in 22m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/786f183acc5e412da3eeb3f75e651e69 : SUCCESS in 17m 42s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/68c8d630a12540dba62dcfa5f464cd6b : SUCCESS in 16m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0952de5288c4803b807672d17e73ef5 : SUCCESS in 12m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb65984a4df94110a085acf109774e97 : SUCCESS in 14m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c3f238c8cc494eb21b57a430f0d42a : SUCCESS in 45m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b98891722be7443cb36757f2954cd7f3 : SUCCESS in 1h 03m 27s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/860c041fa12846c19edc7ae0121e8ee3 : SUCCESS in 56m 16s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/48ae6d995f3747559324d374ce5db4c2 : SUCCESS in 20m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7c8cd270b8e476a9ad76b7423c77658 : FAILURE in 19m 12s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fd8af85ac76f43b0953b6b650c7f2d6e : SUCCESS in 2h 01m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11dd432c872f4b808f26466b78df74e2 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d3152b9f405c68cde905667dd9acb9645326c52f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-16 13:55:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"276ef7e07a7f6ca8fd1db627f6f34a91dae79e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 16:26:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfc7f972c2e54fe983ce3257c9a23c47 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7456bfc484d74475a731d1d0349de559 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd0bcc34eb374f58b39347b547608312 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f07359ff1fbc43db8a334f6bac798b5c : FAILURE in 23m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bbb837fc7761484295e820a9f7cfa3fa : FAILURE in 16m 40s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/9ac30909bea84572b9cb6548198e4d24 : FAILURE in 17m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3557ca5cecbe4878ba99a251806d3a8c : FAILURE in 15m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a50faf660db471fa8e5a87dce4e47ef : FAILURE in 14m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fd1eaae853a43e991785c0979f846c4 : SUCCESS in 45m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/65a62e20f4c64f68afcd078b209a5892 : SUCCESS in 1h 01m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d55b4e67b5a648d3a79f4d536f972673 : SUCCESS in 1h 00m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e3e4c3b782e64888a8261731a77fbe4c : SUCCESS in 18m 20s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4f13a81e8ba44cc2a9af98b8a7c89d1b : FAILURE in 22m 16s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a071f162828240998cea95d654916aed : SUCCESS in 2h 27m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93a37bb58c2347218b7e5c33cbfae59a : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"4a5006a7a5097953593a140b50561795418e69c8","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-16 19:50:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"27be98afd904c04ca565aa5fda1b517d6398f16b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:47:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20cd83ebb43144a6840fa9e93f43ff65 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f43156f9ad041b7bf42d966f10a1753 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61d983f8e54e48d4b279bfe84d50c899 : SUCCESS in 8m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3dbcc78038664314970e1d20c9987c8b : SUCCESS in 16m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d7579f1ec860481494ea9f68252dc85c : SUCCESS in 17m 36s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/46e48619053241d5a1a345a00ece58e8 : SUCCESS in 15m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0cf479bb00f4eb8811c2520f8140ffb : SUCCESS in 13m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30d041124f6a43f8a752dd510cc9b0cd : SUCCESS in 15m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/137d7b06f628468eace9294b4acdd3f8 : SUCCESS in 48m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5c2db84cbc7246a59c1afdfbaeccebbc : SUCCESS in 1h 06m 33s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cbe41aa6eefa4ca8ba3c34eba776f13e : SUCCESS in 58m 08s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fed97b3d66c54839a236059803991bba : SUCCESS in 19m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a736c9abf3340ad988d64f60fc59c35 : FAILURE in 19m 58s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/cc52c26b3a1e4fbd9345e8c9eac9e34b : SUCCESS in 1h 54m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16ffdd4b067d4306bb241d0f9c06cf2a : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":3},{"id":"42b929118c27e8d17ceb15deaa54745360a19822","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-17 06:24:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7187a9c1119b5b640ac3ea723bb5a08842bd9ff7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-17 08:37:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76b115127b228d8521225f1a285752fb2f11db7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:37:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"372ed191fb83d93c1dbff5412d6fd524b19c1e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 10:36:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a28b29ab4b9b44029bd4096d1b025ec2 : SUCCESS in 7m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1fde290fe914a5fbc84a7b10b26d43a : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7173c3f424a4ea0b442555c5560ddc5 : SUCCESS in 5m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c8f85e8b68ce4731afa8c1a2a768e9b4 : SUCCESS in 16m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea535170e21c43cf883dc67c414c947b : SUCCESS in 14m 44s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b244027a6e0d4a348a3870d3e63eafc0 : SUCCESS in 48m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/76da62c14f7c4fb390d0aed90744b8bf : SUCCESS in 1h 04m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c7b217bc4aab48bdbd996f0751b2f1c9 : SUCCESS in 22m 56s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a3685224fe3744bb89aea91c6e89ba01 : SUCCESS in 1h 56m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a6f46a567d4fe48f744056b25e9a16 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0ff1714d80b4657f983a1a0da4f89606bc6a4010","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 10:36:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"820f8f734537d507b1a6ebca88b3dcd7e167a431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 10:40:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ad61ae8f5cb45f4b038858283eac067 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"74f3dcebf3f8ac89d35fec8a0dfbe92468319eda","revisions":{"0ac527497c7aa9ff3b05b47d58f6d9b4b2b1515f":{"kind":"REWORK","_number":1,"created":"2020-04-16 11:32:22.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/720441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/41/720441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/1"}}},"commit":{"parents":[{"commit":"e2a354d877ab35a31b9726281686ee7724b2db95","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e2a354d877ab35a31b9726281686ee7724b2db95"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-15 14:03:18.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-16 11:31:41.000000000","tz":120},"subject":"Add ironic::neutron::dhcpv6_stateful_address_count","message":"Add ironic::neutron::dhcpv6_stateful_address_count\n\nSupport fo additional ip address allocations when using\ndhcpv6 stateful was added in https://review.opendev.org/700002\nand backported to stein in https://review.opendev.org/717205.\n\nThis change adds support in the puppet module to configuring\nthe amount of addresses to allocate.\n\nConflicts:\n    manifests/neutron.pp\n    spec/classes/ironic_neutron_spec.rb\n\nRelated-Bug: #1861032\nCloses-Bug: 1872992\nChange-Id: I09e6ce611b12e92a4a8db64c7f23a491998e956b\n(cherry picked from commit 596abc9587f18b222918a8e413c97fcc1f3affeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/0ac527497c7aa9ff3b05b47d58f6d9b4b2b1515f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/0ac527497c7aa9ff3b05b47d58f6d9b4b2b1515f"}]},"branch":"refs/heads/stable/train"},"d01f1e44abf6eba5ebf85aa79e42aa98b87c1e0b":{"kind":"REWORK","_number":2,"created":"2020-04-16 13:55:48.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/720441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/41/720441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/2"}}},"commit":{"parents":[{"commit":"e2a354d877ab35a31b9726281686ee7724b2db95","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e2a354d877ab35a31b9726281686ee7724b2db95"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-15 14:03:18.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-16 13:55:03.000000000","tz":120},"subject":"Add ironic::neutron::dhcpv6_stateful_address_count","message":"Add ironic::neutron::dhcpv6_stateful_address_count\n\nSupport fo additional ip address allocations when using\ndhcpv6 stateful was added in https://review.opendev.org/700002\nand backported to stein in https://review.opendev.org/717205.\n\nThis change adds support in the puppet module to configuring\nthe amount of addresses to allocate.\n\nConflicts:\n    manifests/neutron.pp\n    spec/classes/ironic_neutron_spec.rb\n\nRelated-Bug: #1861032\nCloses-Bug: 1872992\nChange-Id: I09e6ce611b12e92a4a8db64c7f23a491998e956b\n(cherry picked from commit 596abc9587f18b222918a8e413c97fcc1f3affeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d01f1e44abf6eba5ebf85aa79e42aa98b87c1e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/d01f1e44abf6eba5ebf85aa79e42aa98b87c1e0b"}]},"branch":"refs/heads/stable/train"},"74f3dcebf3f8ac89d35fec8a0dfbe92468319eda":{"kind":"REWORK","_number":3,"created":"2020-04-16 19:50:24.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/41/720441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/41/720441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/41/720441/3"}}},"commit":{"parents":[{"commit":"e2a354d877ab35a31b9726281686ee7724b2db95","subject":"Fix rabbitmq dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/e2a354d877ab35a31b9726281686ee7724b2db95"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-15 14:03:18.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2020-04-16 19:49:39.000000000","tz":120},"subject":"Add ironic::neutron::dhcpv6_stateful_address_count","message":"Add ironic::neutron::dhcpv6_stateful_address_count\n\nSupport fo additional ip address allocations when using\ndhcpv6 stateful was added in https://review.opendev.org/700002\nand backported to stein in https://review.opendev.org/717205.\n\nThis change adds support in the puppet module to configuring\nthe amount of addresses to allocate.\n\nConflicts:\n    manifests/neutron.pp\n    spec/classes/ironic_neutron_spec.rb\n\nRelated-Bug: #1861032\nCloses-Bug: 1872992\nChange-Id: I09e6ce611b12e92a4a8db64c7f23a491998e956b\n(cherry picked from commit 596abc9587f18b222918a8e413c97fcc1f3affeb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/74f3dcebf3f8ac89d35fec8a0dfbe92468319eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/74f3dcebf3f8ac89d35fec8a0dfbe92468319eda"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
