)]}'
{"id":"openstack%2Fpuppet-ironic~721439","triplet_id":"openstack%2Fpuppet-ironic~master~I3a0e85ce6d83001833c829f484baf181a677a5d9","project":"openstack/puppet-ironic","branch":"master","topic":"ipxe_name_base","hashtags":[],"change_id":"I3a0e85ce6d83001833c829f484baf181a677a5d9","subject":"Allow alternative source files for /tftpboot/ipxe.efi","status":"MERGED","created":"2020-04-21 02:08:26.000000000","updated":"2020-04-22 12:07:37.000000000","submitted":"2020-04-22 12:05:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721439-1587557139086-16b58591","meta_rev_id":"476749a1f80f2fb9f15013991b5a2f24bda6dee7","_number":721439,"virtual_id_number":721439,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-04-21 23:24:37.000000000","_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 12:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2020-04-22 08:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-04-22 09:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-22 10:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-22 10:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 02:09:28.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-04-21 02:09:36.000000000","updated_by":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"reviewer":{"_account_id":1926,"name":"Derek Higgins","email":"derekh@redhat.com","username":"derekh"},"state":"REVIEWER"},{"updated":"2020-04-21 02:09:40.000000000","updated_by":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2020-04-22 08:21:16.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2020-04-22 09:09:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-22 10:56:55.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-22 12:05:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa5bdbc6d044c8dbd338ba815b2b5fb44f240985","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 02:08:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca5a7261c6fb640ea155438832047bd800bf869","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 02:15:15.000000000","message":"Patch Set 1:\n\nI\u0027m looking for feedback on this approach to handle a specific UEFI-only environment which needs to also make it possible for ironic-inspector to boot ipxe-snponly-x86_64.efi as the default ipxe efi\n\nAlso something similar needs to be backported to queens","accounts_in_message":[],"_revision_number":1},{"id":"83f6cd8d828ed7d274a466df9e4bc0c22def6f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 03:15:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5a9e938f9f2430d9c117ed40443426f : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a509d68102642128d0258b74655783b : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d14b8f8673724e1ab4ea801857dd775a : SUCCESS in 4m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/226864d04af24910ac512f177d39b36e : FAILURE in 14m 45s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d361599b870743d5a5b05bf539d93489 : FAILURE in 13m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84840a4d26294b289e7b38109cec69b5 : FAILURE in 11m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5188d677cdc4594a1362cca002f9121 : FAILURE in 13m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bca40569259e4c7ebf8c17f055e7a26a : SUCCESS in 49m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e42bce5ad71041a49c4952bc36943ab8 : SUCCESS in 1h 04m 40s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/93d5ef00c51841b5b900f9d7b7111d7a : SUCCESS in 24m 39s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7fa064b0e884402784c40e072da1e685 : FAILURE in 13m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e66704330e9c460c98195b1c3db9fa3f : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce67ceccdd549f999c6d803143d52d36fd4a26fc","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 05:13:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01fe66290a3a1bbcdcd0e41465b717441a592689","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 05:39:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a9e8c094966e7b288de65bacf33d545aa89b2cf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-21 06:42:58.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIt mostly looks good.\n\nCan we add a release note about the new parameter, also ?","accounts_in_message":[],"_revision_number":2},{"id":"a03ac0f57ada6ab325f8adcc8c65683b7d3a1b80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 06:45:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78b834ba7d5c432c94ecf8205519385d : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b59b5218dfd4538aa6f4ec9320b5e07 : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a18de1482cd498e8fbad141454073cc : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ae9ae8a820d847e698fade138f2e61a4 : SUCCESS in 16m 47s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c3364cdabd01448d902248929774a833 : SUCCESS in 16m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa4b6d159d5f42b39d7ff7eafc600d14 : SUCCESS in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8409f9b338548a29e667b15250ea7cd : SUCCESS in 12m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9513116115df466b998e1807efb77f6e : SUCCESS in 52m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cafd19cd092c487cba56f6ac9b2fbe35 : FAILURE in 1h 00m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a27118b8e75c4c87b64d39e52cd8da40 : SUCCESS in 18m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0575c2587d534967ab603eff9b9c8e6a : FAILURE in 8m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c58bb68631f24e038f691335f75648e0 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e36ad234078db68f194c25c6ddabcc33105df43f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-21 08:45:08.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI wonder if simply defaulting to the snponly ipxe is the practical thing to do. Also inline comment, is my opinion.","accounts_in_message":[],"_revision_number":2},{"id":"0e405f2b0484302c2faa8eeaa0feded27378e247","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 21:16:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"417a489d18de5b24d7962198872358197f2df47f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 21:48:01.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Can we add a release note about the new parameter, also ?\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"b39e2c4a7fda0b8751a1908b065234d8695322a1","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 21:49:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db83e538343b354fd2096b94eac73e568296617f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 22:52:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87e4dbadabc6469e9c840536c0635e0c : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8bc01bc07d64a6c92b142212e426d42 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e4a930b04f1489ca608d93cd057e615 : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1ff91507f4314266be08110525dc1d07 : FAILURE in 16m 40s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b659da07330b4da8ac87b788e41494a0 : FAILURE in 18m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b6c280d79e4952b50b2c7e014c8e2d : FAILURE in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7016dae3e31c4340b92cbbe714cb72d1 : FAILURE in 12m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3103fcc8a9b74106a8bbd47ff503a7dc : SUCCESS in 46m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c3b3e9098a3040dd81dff751f720a583 : SUCCESS in 58m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1cb6c965b309413390d94fe2013a6f87 : SUCCESS in 18m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6d4d10e66fe0468c9b63a49e323de9d1 : FAILURE in 8m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b782c13d2d143b980cce2c89dcc6a2a : FAILURE in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc61aecb12da4625b519ec64df26c354 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"4b79220024992296fffc9868ac94ddd4b4cd0b6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-21 23:12:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8eefd39ce855556cf80cf69a448e691e6517bd59","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-04-21 23:24:37.000000000","message":"Patch Set 4:\n\nTakashi, thanks","accounts_in_message":[],"_revision_number":4},{"id":"6e018a317f08afe937360a73eae4e471a6a1c45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:20:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e49b49e4862340caa0ea466b524fba8a : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc2d21aa0fc342bdb4fca8a1567c8d03 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5ffd9e447e347c39e64d88c3ad179e7 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/827d1ced61994bb28082f49151098f4f : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/69ebb4d0f217432ca36db10a99397b83 : SUCCESS in 17m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40d32a06f05f4d9f9791f5bf1b92d4a7 : SUCCESS in 12m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cdd2aa1cceb4453ac910d9b6abbf80e : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3da03634b8cd49739aac14d81cd099b7 : SUCCESS in 58m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6328678309fb435fa3319ad4aea1f18d : SUCCESS in 1h 03m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/21b3a651128a4edaabc0ae36b71a1ba5 : SUCCESS in 19m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/55eeb2a4f01d4a0fa4d62ba3e785a1bf : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02643d3cb1ec46bdab10f31d7ea61809 : FAILURE in 7m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7da63341bdd5491a98e2055c30cc30de : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":4},{"id":"e5a74cc97788fb6ae54372d8d8dd3b24e2645a2e","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2020-04-22 08:21:16.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"199ada2041cf13ccc65c04a18069b3841b9e7efa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-22 09:09:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"52d1af7d9e4a434dcd10d6bd90c99b5c8fb0e693","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-22 10:56:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0c18711f5a1f00eb7ad303f6357a778c16252213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 10:57:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c4c8cf005ad533bdab23182c8e335f6ab5d94d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:05:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d6faef725814b92974de4268456f429 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23ee08a3318746e5a7906528c634dd43 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/471183b37eda44098ec5b5a1cac9f310 : SUCCESS in 6m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e2b62e0fdc37406fb21702793dedd251 : SUCCESS in 16m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5b3fed7b599e4fa9a94e24f4e469c5d4 : SUCCESS in 16m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/277eaa1533684b8e8b8f87de7205bc79 : SUCCESS in 13m 55s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f74e9dc3c2da437f8fd5038d33e1a901 : SUCCESS in 1h 04m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dedf59ccf08c42758878365fb438ba95 : SUCCESS in 1h 06m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e86e20ba30b241448000b85a2cfb8aec : SUCCESS in 23m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0046e9180eb3477e83a831ffc52dfa51 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9d05a8fa1e61ec87f6e5c0f50af6939e952f74e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:05:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c7c0177b049b9e14e2b32e03c1a2d14608bfad00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:07:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24ebc47c495341c4b3cd976b67dbaeb9 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ca47ba7c7f96172723c2a7d00a469986b302694f","revisions":{"f2189f234040c2f3046706ca11c8f5e24eb9f8a7":{"kind":"REWORK","_number":1,"created":"2020-04-21 02:08:26.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/721439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/39/721439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/1"}}},"commit":{"parents":[{"commit":"06547f81a971df7620fb4b0402cd92960e4fd93d","subject":"Add region_name parameter to ironic::drivers::inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/06547f81a971df7620fb4b0402cd92960e4fd93d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 01:59:07.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 02:08:23.000000000","tz":720},"subject":"Allow alternative source files for /tftpboot/ipxe.efi","message":"Allow alternative source files for /tftpboot/ipxe.efi\n\nipxe-bootimgs will soon include the ipxe efi variant\nipxe-snponly-x86_64.efi which has no networking stack, relying instead\non the UEFI networking stack. This image is required in some UEFI\nenvironments instead of the full ipxe-x86_64.efi, so there needs to be\na mechanism for serving it by default.\n\nThis change makes this possible by setting the puppet parameter\n$::ironic::pxe::common::ipxe_name_base to ipxe-snponly.\n\nChange-Id: I3a0e85ce6d83001833c829f484baf181a677a5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f2189f234040c2f3046706ca11c8f5e24eb9f8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/f2189f234040c2f3046706ca11c8f5e24eb9f8a7"}]},"branch":"refs/heads/master"},"0723e659e5b4814ad4a042c251d2faed273a10fa":{"kind":"REWORK","_number":2,"created":"2020-04-21 05:39:51.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/721439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/39/721439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/2"}}},"commit":{"parents":[{"commit":"06547f81a971df7620fb4b0402cd92960e4fd93d","subject":"Add region_name parameter to ironic::drivers::inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/06547f81a971df7620fb4b0402cd92960e4fd93d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 01:59:07.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 05:39:20.000000000","tz":720},"subject":"Allow alternative source files for /tftpboot/ipxe.efi","message":"Allow alternative source files for /tftpboot/ipxe.efi\n\nipxe-bootimgs will soon include the ipxe efi variant\nipxe-snponly-x86_64.efi which has no networking stack, relying instead\non the UEFI networking stack. This image is required in some UEFI\nenvironments instead of the full ipxe-x86_64.efi, so there needs to be\na mechanism for serving it by default.\n\nThis change makes this possible by setting the puppet parameter\nironic::pxe::ipxe_name_base to ipxe-snponly.\n\nChange-Id: I3a0e85ce6d83001833c829f484baf181a677a5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/0723e659e5b4814ad4a042c251d2faed273a10fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/0723e659e5b4814ad4a042c251d2faed273a10fa"}]},"branch":"refs/heads/master"},"34e046b1d2d200f471d3e42ffaf76f9340d8dc8e":{"kind":"REWORK","_number":3,"created":"2020-04-21 21:49:47.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/39/721439/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/39/721439/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/3"}}},"commit":{"parents":[{"commit":"06547f81a971df7620fb4b0402cd92960e4fd93d","subject":"Add region_name parameter to ironic::drivers::inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/06547f81a971df7620fb4b0402cd92960e4fd93d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 01:59:07.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 21:49:35.000000000","tz":720},"subject":"Allow alternative source files for /tftpboot/ipxe.efi","message":"Allow alternative source files for /tftpboot/ipxe.efi\n\nipxe-bootimgs will soon include the ipxe efi variant\nipxe-snponly-x86_64.efi which has no networking stack, relying instead\non the UEFI networking stack. This image is required in some UEFI\nenvironments instead of the full ipxe-x86_64.efi, so there needs to be\na mechanism for serving it by default.\n\nThis change makes this possible by setting the puppet parameter\nironic::pxe::ipxe_name_base to ipxe-snponly.\n\nChange-Id: I3a0e85ce6d83001833c829f484baf181a677a5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/34e046b1d2d200f471d3e42ffaf76f9340d8dc8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/34e046b1d2d200f471d3e42ffaf76f9340d8dc8e"}]},"branch":"refs/heads/master"},"ca47ba7c7f96172723c2a7d00a469986b302694f":{"kind":"REWORK","_number":4,"created":"2020-04-21 23:12:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/721439/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/39/721439/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/39/721439/4"}}},"commit":{"parents":[{"commit":"06547f81a971df7620fb4b0402cd92960e4fd93d","subject":"Add region_name parameter to ironic::drivers::inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/06547f81a971df7620fb4b0402cd92960e4fd93d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-04-21 01:59:07.000000000","tz":720},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-21 23:12:04.000000000","tz":0},"subject":"Allow alternative source files for /tftpboot/ipxe.efi","message":"Allow alternative source files for /tftpboot/ipxe.efi\n\nipxe-bootimgs will soon include the ipxe efi variant\nipxe-snponly-x86_64.efi which has no networking stack, relying instead\non the UEFI networking stack. This image is required in some UEFI\nenvironments instead of the full ipxe-x86_64.efi, so there needs to be\na mechanism for serving it by default.\n\nThis change makes this possible by setting the puppet parameter\nironic::pxe::ipxe_name_base to ipxe-snponly.\n\nChange-Id: I3a0e85ce6d83001833c829f484baf181a677a5d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/ca47ba7c7f96172723c2a7d00a469986b302694f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/ca47ba7c7f96172723c2a7d00a469986b302694f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
