)]}'
{"id":"openstack%2Fpuppet-ironic~723969","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fussuri~I5c4350bd033fa4a0e18d3063acb3a33be5cc1ac9","project":"openstack/puppet-ironic","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I5c4350bd033fa4a0e18d3063acb3a33be5cc1ac9","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-28 14:53:09.000000000","updated":"2020-04-29 02:52:18.000000000","submitted":"2020-04-29 02:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723969-1588128738844-e686b5f9","meta_rev_id":"62be6f8decfa9fc055504f94794730686566b3ce","_number":723969,"virtual_id_number":723969,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 02:52:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-04-28 16:45:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:45:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:46:58.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 02:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2203d2f3fc0fbb2c44ed48c6288e608976bfbd98","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:53:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"341295565e36fcc0469fd1731f56615d1f00c416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:21:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/671d00fae1cd49ed8e85e82f3aca447f : SUCCESS in 7m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2159e1dc93a487296ee5f2ccbe08f5e : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67879d83bd14485e8c7ff6047ea82230 : SUCCESS in 5m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/24071cf17c464332b774ba13fb3adbbc : SUCCESS in 19m 31s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/91553dbd5a7244208e4d61d81e33ee65 : SUCCESS in 21m 04s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffeb6b7c570744ce96df694c1cc6a106 : SUCCESS in 18m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f547cede6fd8431c86c1687f098fb258 : SUCCESS in 15m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c2ca21aae5428c9ef5c4921caef207 : SUCCESS in 56m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7224149e51b94e34a9a1e17a57a36ba2 : SUCCESS in 1h 09m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a67948780fcf44189e044203d6f6f6bd : SUCCESS in 22m 43s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0e621983b3aa4dd98309ab58f890cd8b : FAILURE in 11m 57s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc5da1f5d6a84c87ad4c3dd144f54240 : FAILURE in 9m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad38426400b4fad4ddfe6eae853c71e2be6de7a8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:45:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"68f300dd4a5b61fdcf308dc06cc116a5fa7899e1","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:46:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4c3c6ef932ec3ec4080cbe4d69041d1e554160c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:47:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3076094d67f0001bb428314c73c6ebdee6c7a8b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:52:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5407236913aea2f1821417e93a5c526bf03ad466","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:52:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad4df541ae4d4f17a917754a7b401d0e : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf5cb4f8d07e47b9a5224c4ddbe34e28 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6e90e43ecd14798a14933e134ebe1ff : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/602800a3bf4b4a15a334002d31dbc5a8 : SUCCESS in 16m 25s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/397140d082f941549c99c0ecc24d905c : SUCCESS in 14m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e78ba74a55874cb196d7fe0d16f0d961 : SUCCESS in 12m 15s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/920052c385c143ce8f55da5d28c97d71 : SUCCESS in 51m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1b63cd56f4c14689b0caf3b36640c875 : SUCCESS in 1h 00m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1b5406924af74dd3b96edb277d0fdd6f : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e860b17ce389ca5af6205dbcad3aff8d755f7a9","revisions":{"9e860b17ce389ca5af6205dbcad3aff8d755f7a9":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:53:09.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/69/723969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/69/723969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/69/723969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/69/723969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/69/723969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/69/723969/1"}}},"commit":{"parents":[{"commit":"a01451dd1403ce5032ab4654a35bf7402fb0b154","subject":"Prepare Ussuri RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/a01451dd1403ce5032ab4654a35bf7402fb0b154"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:53:07.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:53:07.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I5c4350bd033fa4a0e18d3063acb3a33be5cc1ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9e860b17ce389ca5af6205dbcad3aff8d755f7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9e860b17ce389ca5af6205dbcad3aff8d755f7a9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
