)]}'
{"id":"openstack%2Fpuppet-ironic~723970","triplet_id":"openstack%2Fpuppet-ironic~stable%2Fussuri~I3456e8fd7e3a24d4dd60454f11c934232fe399cf","project":"openstack/puppet-ironic","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I3456e8fd7e3a24d4dd60454f11c934232fe399cf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:53:12.000000000","updated":"2020-05-01 08:26:46.000000000","submitted":"2020-05-01 08:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723970-1588321531400-79c713a5","meta_rev_id":"bdf780ac8fbaf1a878ad9425443114e8981669e6","_number":723970,"virtual_id_number":723970,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-04-30 04:56:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-01 08:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:47:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-28 16:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:47:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:45:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:47:04.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-30 04:56:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-01 08:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"837b6ad111e3694eca9ac2821f6de60178f17b25","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:53:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80949cdd28500c853b18fdf00209f7087908b8a7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:45:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fcaab7582dc6b9238fc3f9ec739ceae6b5e6504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:34:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/479806498fce423cb797b5d46d9f5ef1 : SUCCESS in 7m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cab7c8af02240699a5f74944804a1f7 : SUCCESS in 7m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/553798d1d2e84e0f8f6f589ec9e2434b : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7b231943b11e40619eb7d736d2d5dd0e : SUCCESS in 16m 02s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5f196ab2538a49fa9692d040117c3c5a : SUCCESS in 16m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/794f7258334c41f495ae27146f77d8e0 : SUCCESS in 15m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc87bf3764df40bc83fdf1d5eab782fa : SUCCESS in 16m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e6baa07fd8a4093a22ef0243c8d99ba : SUCCESS in 57m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/16c1508ada82456e8effbda7c1295767 : SUCCESS in 1h 34m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c4bad44cb9084d9f98ddbd2d4d2b79b2 : SUCCESS in 23m 37s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/52d85a191f27402eb4b12d915891290f : FAILURE in 13m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fac5c7c008214edfaf8e2c77de80fe94 : FAILURE in 11m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff0392ad4ef641ee87b1bb1ff0851be3 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6cf7ff70836246abecd5d56c3414cf2fb686a0b9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:47:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad7445cb1466b3d1395251480495eaf505d32195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23b152357821c48df9917de25512fb3f6258e79c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 03:29:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa717706bebe4963aa1ef6e2410439c2 : SUCCESS in 7m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27e2e593819343fba0b01095772cb1ed : SUCCESS in 5m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4acbef9b70b04a08aa16b54c2fa963e4 : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b6e8eae120f84e6abcab8809026c3499 : SUCCESS in 18m 30s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b6fb43ed17f64adf85dd2080adb5b9f3 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc70ef7b0fd3475eb05d185fb9eae23b : SUCCESS in 11m 02s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cb569cf97414bb2ae416e04deece423 : SUCCESS in 1h 40m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb8c9a13d8104a9bbf604351f5916457 : FAILURE in 1h 22m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cc2080f8db4444f3a64bd04177eda34e : SUCCESS in 22m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6804e92308e41828737e11f58ed7319 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"13a4bb09d657db605fbdec9d4842fe1a74096690","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:16:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27f04d548b95a1cba6b2b7a73b8adb0cc04f1826","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:31:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d89ed9cbfc38c6cb0a2f4c990c1a92b338c33ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:41:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fed0c2b640744142b0b58139db9446c5 : SUCCESS in 6m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0862380a28464558ade0c8e6d5f6b4b1 : SUCCESS in 6m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/139190a98d4a44e997ea127913b1db66 : SUCCESS in 6m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/acc0e44f6e0743bf87ba10c78ea1e71f : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c872cdeb405e415cbeb12b0eadf9b51c : SUCCESS in 16m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65686a6ad7ce4bcaa9ee150870b2c652 : SUCCESS in 15m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/331fa502ea5842d59dc40126bbd9cc4d : SUCCESS in 16m 13s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/680a8e8e7b8c4e73ba2b57609f1477e4 : SUCCESS in 56m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b18721c3b19948538cbc30a65b504030 : FAILURE in 39m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fdbc3ec8003f4f0dae1075bd6592deba : SUCCESS in 20m 28s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/75b61dbbc6234bbd92f45c3651fbfcc5 : FAILURE in 12m 35s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c0b90d50ea4701b0b3cd114271deeb : FAILURE in 7m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f9e7cf3551b424aa1b420ea39e72b66 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4295ad70da1be539e6a241e0cdb5230d275d78a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 16:03:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"feb798010a0371f4b6fa24fc92bd22d32dcf12c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:17:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21b7dc70c82640af999d68e43aa56d0b : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f886640eeb04e75b6bfee4531341dac : SUCCESS in 6m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e170ed3ee0d4f0fb1ecc1c5e5d0d99f : SUCCESS in 8m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d573e83907cf466dbf1422f9ea305469 : SUCCESS in 18m 24s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2e894cd7c0a544ab884b6eeeec36ac6a : SUCCESS in 23m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fcfc05bcf02431784e99562471d6f14 : SUCCESS in 15m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd3cfd1898f2411b8d2a93e1d74c8072 : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97e15f0923d248fd89c3623b78768bea : SUCCESS in 1h 09m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9548165524e14090a00d77ac8bdac398 : FAILURE in 39m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7f88be2000434321a2aad6998ce3c124 : SUCCESS in 29m 55s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a1b23fee0e0f4c26a03c40bd5d7c5107 : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b3d9644d3764afd930c18260c0f9ea3 : FAILURE in 11m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526ef1451fd74af48f29ea85ede24dbc : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fe8f7509d6e724acff571c243a20b547a8373121","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 04:56:04.000000000","message":"Patch Set 1:\n\nNow we need https://review.opendev.org/#/c/724312/ to fix broken scenario002.","accounts_in_message":[],"_revision_number":1},{"id":"f56d30ac0b5087a9ab99e48d1fb853b97d6b03c3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-01 06:10:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27a6d9963327d6b297c93d8538c41610514cd652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 07:19:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/447f73a27f454b4d8f8fb796d23bdfa1 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e35c881e0994f7a86f56af063864af1 : SUCCESS in 4m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e74dea56e0746d3b699850c191741d3 : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a53208180edd45a7897cfd6e81d57c57 : SUCCESS in 21m 23s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f5886637142f4dabbba69301a41968cd : SUCCESS in 13m 48s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06408763852d48a8827f752dc6d631b5 : SUCCESS in 14m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c890ca7bbf8044e28f39f72ab15073d5 : SUCCESS in 14m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1690628e0dd14e8e900d26497778db33 : SUCCESS in 58m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1d8d4d39f9cf4c3cbc09746e6ac65a61 : SUCCESS in 1h 06m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1f5b7132f7844a0b9066f9a4551ca98a : SUCCESS in 24m 31s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9729aa030d4546a99d78924696e9a7e5 : FAILURE in 12m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55aab3d44c0d4432b47779563537b789 : FAILURE in 10m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb1ff9844f6f464cb92f4bb3e3cd9452 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"443ff7dbcd1b46f1195e0702fd1f55d69fd4fbc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 07:20:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa15a8275dd477ac3bf7aa4e79fc959614499c12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 08:25:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdc52ec6981f4e68aebb8fae8449da5f : SUCCESS in 5m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4de5ef8a54324ec3adc7f19dd99e14a8 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fe83743c4b34a549b2bc2ec0ab63e06 : SUCCESS in 4m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/103afcf319ce41a6bfe8f8a97764cb5b : SUCCESS in 12m 19s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f0c54abfb58a4d11bd3e5c9a3746f273 : SUCCESS in 13m 57s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b90d9aeeb8514706b1f8998442776d23 : SUCCESS in 11m 24s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47871f2b2be7442ba272c29fcf90a9c0 : SUCCESS in 54m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1572f3dd82a1456fad386acf2f259b40 : SUCCESS in 1h 03m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9536426cba1a4c5da8d19d55d6f9c37c : SUCCESS in 19m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45392caf3b2340b4839ca482924586fe : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bfb7b45d4514393173244003851f6e6563551016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 08:25:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"25666ab80ecee2c99a64b5cde1a27779170d0cb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 08:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05acad9d4e8e40e2b14bbfab77e1fd45 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3972508bbf8369cd14d8837a23cea2e68db6f5a","revisions":{"b3972508bbf8369cd14d8837a23cea2e68db6f5a":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:53:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/70/723970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/70/723970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/70/723970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/70/723970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/70/723970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/70/723970/1"}}},"commit":{"parents":[{"commit":"9e860b17ce389ca5af6205dbcad3aff8d755f7a9","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/9e860b17ce389ca5af6205dbcad3aff8d755f7a9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:53:10.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:53:10.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I3456e8fd7e3a24d4dd60454f11c934232fe399cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b3972508bbf8369cd14d8837a23cea2e68db6f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/b3972508bbf8369cd14d8837a23cea2e68db6f5a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
