)]}'
{"id":"openstack%2Fpuppet-ironic~748083","triplet_id":"openstack%2Fpuppet-ironic~master~If3d204f6807937cacbb9e8cc541cc3c230b1b670","project":"openstack/puppet-ironic","branch":"master","topic":"bug/1892952","hashtags":[],"change_id":"If3d204f6807937cacbb9e8cc541cc3c230b1b670","subject":"Add mysql_enable_ndb option","status":"MERGED","created":"2020-08-26 03:54:20.000000000","updated":"2020-08-29 07:36:58.000000000","submitted":"2020-08-29 07:35:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748083-1598686547688-46488824","meta_rev_id":"73fd47f737228e45b80e76ff9bd63eb428e9891f","_number":748083,"virtual_id_number":748083,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-29 04:28:54.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-29 07:35:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 12:47:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-08-26 12:41:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-26 15:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 12:47:33.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-26 15:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 12:41:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-26 12:47:33.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 15:46:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-29 04:28:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-29 07:35:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c19d169eac0865c45d765103d0ddc3ebbcce483d","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afe19c43b894a379e915f5d1cd98e5d9b21520ea","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:57:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cd7dd4171cd38e05c37295997dc6afb760258a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 05:47:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dead78070294ed0b4425e525acb845b : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5930870519546e79a1976d29779b8a8 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31b56f8822524c7a9f17137f9f7d3226 : SUCCESS in 4m 12s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4d2640cc52c342f9b81be74cf789fe0c : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6a884b3580054c52a3096c7bca7981b6 : SUCCESS in 14m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8b5fe3c81774b94941fa9686fc74fec : SUCCESS in 11m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8696cc73a4074426afc63b6a514f7e2b : SUCCESS in 13m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33a863cd278c4cfcad8281fed16a042c : FAILURE in 47m 49s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1b4b916d2560439ea531085df315c81e : SUCCESS in 1h 30m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/643a59fe07094c4c8982b2fb9d43af7f : FAILURE in 9m 05s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/865bf5d1f5404d979c189c0a093e81e9 : FAILURE in 7m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7ebaab953a744a6849c7fa83117d0d0 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9f2e4fc42ac2bb98e57bbbfa0679071fad53050b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-26 12:41:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e6082cfef3b7b19482b3725d2f6bb3017165a5f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 12:47:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"468cdc316bdbb0fb3f655f3fddef8df3e80fc4eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-26 15:46:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d26a5f2dc52b0e959abc0734aead5836f0e04617","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 15:47:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f87785e330352c8be362811dbdb69ed30bee60ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 17:29:23.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4391df24d30340dd88718d372626d137 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ba652edf2184ba8973ca4864d66af8f : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c21aa70e1e874d22a4868e64f51dcea3 : SUCCESS in 3m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c29e8a485fbe4c239bf3db2382d513c0 : SUCCESS in 10m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5d977b7c57ab4404bc9aa5494b6d06bb : SUCCESS in 14m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0601d5603c1c4a63b764ab4b53d0cc3a : SUCCESS in 11m 15s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ccbf12ca9724f8ca7e804b8919e9e47 : FAILURE in 53m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2b0c281c4a3348039da3599d4821f16f : SUCCESS in 1h 31m 43s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ff3e7a022b43428cb9550a4dcfd78c20 : FAILURE in 19m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cff98a53102440b3965acef144e4aeb1 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8def58cb50828f0973594c71625630dfd3c4c946","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 14:18:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"67087100adf68cd6c996bcb817bf79c08fffa050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 15:59:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/715a52ce791f415daeb9f0a610ec16b6 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37dff38963854c73992e4b26657eac11 : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/615bf5a7733e4b928714cf4613787b05 : SUCCESS in 3m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/55908cf72c2247eaa43c17a320b1672b : SUCCESS in 13m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bbdbfa18a29f4a3d8b113712ebc2a30f : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01c4925049f44306b32d018f576c562c : SUCCESS in 10m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ce479710a4e4a878746cfba2a780fed : SUCCESS in 13m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/208b3037e7f74c76897cfaf6fffdd3bb : FAILURE in 1h 00m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ea676ece04734ca185a6eba3cad81910 : SUCCESS in 1h 27m 49s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4f4fc6b467c54cb9afca13ed69458965 : FAILURE in 19m 12s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdf6f0a0a6b6497b8d00578b104a1afc : SUCCESS in 17m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffd5f6a38ca841b48ad09774a91101b6 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4265b586b1f20999872081888a0b2524e92fac3c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-28 10:07:23.000000000","message":"Patch Set 2:\n\nWe need https://review.opendev.org/#/c/748479/ to fix broken centos8 beaker.","accounts_in_message":[],"_revision_number":2},{"id":"3c4f2478eacdabc149bbe6acb9f72f320ad82707","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-29 04:28:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd4df1e1882654adf7d0735d8b4708eb3127095c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 06:03:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f0843bc1aa74470b97afae46057e53a : SUCCESS in 3m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee912035512148ada3df09f89ff15146 : SUCCESS in 2m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8a2d3b814f544b9b4077841a2da29a2 : SUCCESS in 2m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bb2fa19deaab497b862d963a6101b344 : SUCCESS in 14m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2a5a688b47244671ada222d8bc6c9a69 : SUCCESS in 11m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2966291e06954d19a77d36020d74f6d1 : SUCCESS in 8m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b33755b6321446783e3431e98b3238b : SUCCESS in 10m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d4c00311e474b47848e6deef6cb0a7f : FAILURE in 48m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d1c66d5b43814fea80ccc1cd2712bded : SUCCESS in 1h 28m 37s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8e723dbd780042fbb0548a125adb7fd3 : SUCCESS in 17m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5177e71499034675af34e292e467aa5e : SUCCESS in 13m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1f3d9f755114427b30d8844a89f9f10 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"fcfb8446ef0471e3f658c224a52b62085cc065b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 06:03:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c52e974d28c65a910c12997c5d628b064f2bfc05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:35:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c3acf8bc072b2c402a9d8bb7795a0d8fc5133d00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:35:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cc0416b2b0f4a0484a5d293bca56ad7 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/005fb85339f545348c8d9c2d406d3eee : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4f6aba9006d4bd0916f4939a9089fdf : SUCCESS in 3m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5c7a34a6e6904ab38f5c6d4053e000f6 : SUCCESS in 14m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/644bd1dbb13c43c69cb8a614afdd89c8 : SUCCESS in 12m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f42328ee37e34083914f23cf0e0c1251 : SUCCESS in 9m 33s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbe28d54683a45c1ab2cf8ebc79dcbc3 : FAILURE in 1h 22m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/214320b7439d45f2966729b0ca16a7c8 : SUCCESS in 1h 26m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b8c35302367d47b0b2883b9c8610e945 : SUCCESS in 23m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2451a28a1e224b12a6b3bfca9ac6f9ab : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"0c94974795295beea257bf79bed20706cacce89e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:36:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5290bbdbd1994f658ad746c0761389ba : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"841f27fc03fb0d9f351352d7284fc51902233d0a","revisions":{"002af7aead84d48f662b1b94fd3d10dc62957ab4":{"kind":"REWORK","_number":1,"created":"2020-08-26 03:54:20.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/83/748083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/748083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/1"}}},"commit":{"parents":[{"commit":"3f682aa5dd34afaebf5599f5e1e504dd83b9f2ae","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3f682aa5dd34afaebf5599f5e1e504dd83b9f2ae"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: If3d204f6807937cacbb9e8cc541cc3c230b1b670\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/002af7aead84d48f662b1b94fd3d10dc62957ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/002af7aead84d48f662b1b94fd3d10dc62957ab4"}]},"branch":"refs/heads/master"},"841f27fc03fb0d9f351352d7284fc51902233d0a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-26 03:57:32.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/83/748083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ironic","ref":"refs/changes/83/748083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ironic refs/changes/83/748083/2"}}},"commit":{"parents":[{"commit":"3f682aa5dd34afaebf5599f5e1e504dd83b9f2ae","subject":"Expose vhost_custom_fragment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/3f682aa5dd34afaebf5599f5e1e504dd83b9f2ae"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:55:59.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: If3d204f6807937cacbb9e8cc541cc3c230b1b670\nDepends-On: https://review.opendev.org/#/c/748067\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/841f27fc03fb0d9f351352d7284fc51902233d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ironic/commit/841f27fc03fb0d9f351352d7284fc51902233d0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
